Changing auth to use implicit environment. - #419
Conversation
This comment was marked as spam.
This comment was marked as spam.
Sorry, something went wrong.
This comment was marked as spam.
This comment was marked as spam.
Sorry, something went wrong.
|
LGTM. Tough that the regression tests don't run on Travis until after the merge, though. |
|
Tough but worth it for the speed-up. I ran them lots and lots though. I was actually planning on letting them fail once before the new environment variables are set up and then re-running after. WDYT? |
|
I just pushed 1a1dac0, so we can wait until that passes before merging. |
|
"Fire when ready, Gridley!" |
This comment was marked as spam.
This comment was marked as spam.
Sorry, something went wrong.
|
@silvolu addressed the incorrect description in 78adc96 This fixes #247 though that is the very issue @craigcitro and I are trying to sort out in googleapis/oauth2client#93. #335 is in some sense already done but I still plan to
This does not address #337 in total. I wanted things like dataset.get_dataset()to work without having to pass in a dataset ID if one could be implied from the environment i.e. via GCE or GAE specific environ vars / metadata or via an environment variable. |
|
Feel free to merge if there is nothing else to discuss. I am grading a final right now so am a bit MIA. |
|
Maybe better not merge until we can figure why the trunk has broken regression tests. Of course, they are passing in my sandbox. Looks like @silvolu kicked off a re-run? |
|
@tseaver I did, this one looks extremely transient :) |
|
Build is back to green, merging this one. |
Changing auth to use implicit environment.
|
There's a decryption error on the credentials, I think @dhermes encrypted the file in his fork. I'll update the credentials file. |
|
See my previous comment. The |
|
@dhermes I've set that variable. |
|
Green makes Silvano happy: |
|
Thanks @silvolu. We are green! w00t |
* chore: Update gapic-generator-python to v1.8.4 PiperOrigin-RevId: 507808936 Source-Link: googleapis/googleapis@64cf849 Source-Link: googleapis/googleapis-gen@53c48ca Copy-Tag: eyJwIjoiLmdpdGh1Yi8uT3dsQm90LnlhbWwiLCJoIjoiNTNjNDhjYWMxNTNkM2IzN2YzZDJjMmRlYzQ4MzBjZmQ5MWVjNDE1MyJ9 * 🦉 Updates from OwlBot post-processor See https://github.com/googleapis/repo-automation-bots/blob/main/packages/owl-bot/README.md --------- Co-authored-by: Owl Bot <gcf-owl-bot[bot]@users.noreply.github.com>
#419) chore: Update test-samples-impl.sh python3.6 --> python3.9 (#418) Source-Link: googleapis/synthtool@1f07110 Post-Processor: gcr.io/cloud-devrel-public-resources/owlbot-python:latest@sha256:a5d81b61dfd1a432d3c03f51a25d2e71b37be24da509966d50724aea7c57c5c2 Co-authored-by: Owl Bot <gcf-owl-bot[bot]@users.noreply.github.com> Co-authored-by: nicain <nicholascain@google.com> fix test
Source-Link: https://togithub.com/googleapis/synthtool/commit/909573ce9da2819eeb835909c795d29aea5c724e Post-Processor: gcr.io/cloud-devrel-public-resources/owlbot-python:latest@sha256:ddf4551385d566771dc713090feb7b4c1164fb8a698fe52bbe7670b24236565b
Source-Link: googleapis/synthtool@7197a00 Post-Processor: gcr.io/cloud-devrel-public-resources/owlbot-python:latest@sha256:c43f1d918bcf817d337aa29ff833439494a158a0831508fda4ec75dc4c0d0320 Co-authored-by: Owl Bot <gcf-owl-bot[bot]@users.noreply.github.com>
[](https://renovatebot.com) This PR contains the following updates: | Package | Type | Update | Change | Age | Adoption | Passing | Confidence | |---|---|---|---|---|---|---|---| | [actions/setup-python](https://togithub.com/actions/setup-python) | action | major | `v3` -> `v4` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | | [google-cloud-bigquery](https://togithub.com/googleapis/python-bigquery) | | patch | `==3.3.0` -> `==3.3.1` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | | [protobuf](https://developers.google.com/protocol-buffers/) | | major | `>= 3.19.0, <4.0.0dev` -> `>=4.21.4, <4.22.0` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | --- ### Release Notes <details> <summary>actions/setup-python</summary> ### [`v4`](https://togithub.com/actions/setup-python/compare/v3...v4) [Compare Source](https://togithub.com/actions/setup-python/compare/v3...v4) </details> <details> <summary>googleapis/python-bigquery</summary> ### [`v3.3.1`](https://togithub.com/googleapis/python-bigquery/blob/HEAD/CHANGELOG.md#​331-httpsgithubcomgoogleapispython-bigquerycomparev330v331-2022-08-09) ##### Bug Fixes - **deps:** allow pyarrow < 10 ([#​1304](https://togithub.com/googleapis/python-bigquery/issues/1304)) ([13616a9](https://togithub.com/googleapis/python-bigquery/commit/13616a910ba2e9b7bc3595847229b56e70c99f84)) </details> --- ### Configuration 📅 **Schedule**: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined). 🚦 **Automerge**: Disabled by config. Please merge this manually once you are satisfied. ♻ **Rebasing**: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox. 👻 **Immortal**: This PR will be recreated if closed unmerged. Get [config help](https://togithub.com/renovatebot/renovate/discussions) if that's undesired. --- - [ ] <!-- rebase-check -->If you want to rebase/retry this PR, click this checkbox. --- This PR has been generated by [Mend Renovate](https://www.mend.io/free-developer-tools/renovate/). View repository job log [here](https://app.renovatebot.com/dashboard#github/googleapis/python-dlp). <!--renovate-debug:eyJjcmVhdGVkSW5WZXIiOiIzMi4xNTAuNCIsInVwZGF0ZWRJblZlciI6IjMyLjE1MC40In0=-->
Co-authored-by: release-please[bot] <55107282+release-please[bot]@users.noreply.github.com>
Co-authored-by: Anthonios Partheniou <partheniou@google.com>
* feat: Add routing information in Cloud Build GRPC clients PiperOrigin-RevId: 547283196 Source-Link: googleapis/googleapis@64d54ff Source-Link: googleapis/googleapis-gen@1de1822 Copy-Tag: eyJwIjoiLmdpdGh1Yi8uT3dsQm90LnlhbWwiLCJoIjoiMWRlMTgyMjExMDliNGEwMjVjMGIyZThjMjUzMjRjYjAxYjZkZjYyNiJ9 * 🦉 Updates from OwlBot post-processor See https://github.com/googleapis/repo-automation-bots/blob/main/packages/owl-bot/README.md * fix: Revert breaking change of main pattern PiperOrigin-RevId: 547506413 Source-Link: googleapis/googleapis@8080bbe Source-Link: googleapis/googleapis-gen@ac98388 Copy-Tag: eyJwIjoiLmdpdGh1Yi8uT3dsQm90LnlhbWwiLCJoIjoiYWM5ODM4OGJiODMyZGJlYmMwYmNkNmI3NDgwMDEwNTczYzBiYmRjNCJ9 * 🦉 Updates from OwlBot post-processor See https://github.com/googleapis/repo-automation-bots/blob/main/packages/owl-bot/README.md --------- Co-authored-by: Owl Bot <gcf-owl-bot[bot]@users.noreply.github.com>
#419) * chore(deps): update all dependencies * revert changes and ignore future changes for environment specific pins Co-authored-by: Anthonios Partheniou <partheniou@google.com>
#419) * chore(deps): update all dependencies * revert changes and ignore future changes for environment specific pins Co-authored-by: Anthonios Partheniou <partheniou@google.com>
[](https://renovatebot.com) This PR contains the following updates: | Package | Change | Age | Adoption | Passing | Confidence | |---|---|---|---|---|---| | [google-cloud-bigquery](https://togithub.com/googleapis/python-bigquery) | `==2.30.1` -> `==2.31.0` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | | [google-cloud-bigquery-storage](https://togithub.com/googleapis/python-bigquery-storage) | `==2.10.0` -> `==2.10.1` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | | [google-cloud-testutils](https://togithub.com/googleapis/python-test-utils) | `==1.1.0` -> `==1.3.1` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | | [pandas](https://pandas.pydata.org) ([source](https://togithub.com/pandas-dev/pandas)) | `==1.3.4` -> `==1.3.5` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | | [pyarrow](https://arrow.apache.org/) | `==6.0.0` -> `==6.0.1` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | --- ### Release Notes <details> <summary>googleapis/python-bigquery</summary> ### [`v2.31.0`](https://togithub.com/googleapis/python-bigquery/blob/HEAD/CHANGELOG.md#​2310-httpswwwgithubcomgoogleapispython-bigquerycomparev2301v2310-2021-11-24) [Compare Source](https://togithub.com/googleapis/python-bigquery/compare/v2.30.1...v2.31.0) ##### Features - allow cell magic body to be a $variable ([#​1053](https://www.togithub.com/googleapis/python-bigquery/issues/1053)) ([3a681e0](https://www.github.com/googleapis/python-bigquery/commit/3a681e046819df18118aa0b2b5733416d004c9b3)) - promote `RowIterator.to_arrow_iterable` to public method ([#​1073](https://www.togithub.com/googleapis/python-bigquery/issues/1073)) ([21cd710](https://www.github.com/googleapis/python-bigquery/commit/21cd71022d60c32104f8f90ee2ca445fbb43f7f3)) ##### Bug Fixes - apply timeout to all resumable upload requests ([#​1070](https://www.togithub.com/googleapis/python-bigquery/issues/1070)) ([3314dfb](https://www.github.com/googleapis/python-bigquery/commit/3314dfbed62488503dc41b11e403a672fcf71048)) ##### Dependencies - support OpenTelemetry >= 1.1.0 ([#​1050](https://www.togithub.com/googleapis/python-bigquery/issues/1050)) ([4616cd5](https://www.github.com/googleapis/python-bigquery/commit/4616cd58d3c6da641fb881ce99a87dcdedc20ba2)) ##### [2.30.1](https://www.github.com/googleapis/python-bigquery/compare/v2.30.0...v2.30.1) (2021-11-04) ##### Bug Fixes - error if eval()-ing repr(SchemaField) ([#​1046](https://www.togithub.com/googleapis/python-bigquery/issues/1046)) ([13ac860](https://www.github.com/googleapis/python-bigquery/commit/13ac860de689ea13b35932c67042bc35e388cb30)) ##### Documentation - show gcloud command to authorize against sheets ([#​1045](https://www.togithub.com/googleapis/python-bigquery/issues/1045)) ([20c9024](https://www.github.com/googleapis/python-bigquery/commit/20c9024b5760f7ae41301f4da54568496922cbe2)) - use stable URL for pandas intersphinx links ([#​1048](https://www.togithub.com/googleapis/python-bigquery/issues/1048)) ([73312f8](https://www.github.com/googleapis/python-bigquery/commit/73312f8f0f22ff9175a4f5f7db9bb438a496c164)) </details> <details> <summary>googleapis/python-bigquery-storage</summary> ### [`v2.10.1`](https://togithub.com/googleapis/python-bigquery-storage/blob/HEAD/CHANGELOG.md#​2101-httpswwwgithubcomgoogleapispython-bigquery-storagecomparev2100v2101-2021-11-11) [Compare Source](https://togithub.com/googleapis/python-bigquery-storage/compare/v2.10.0...v2.10.1) </details> <details> <summary>googleapis/python-test-utils</summary> ### [`v1.3.1`](https://togithub.com/googleapis/python-test-utils/blob/HEAD/CHANGELOG.md#​131-httpswwwgithubcomgoogleapispython-test-utilscomparev130v131-2021-12-07) [Compare Source](https://togithub.com/googleapis/python-test-utils/compare/v1.3.0...v1.3.1) ### [`v1.3.0`](https://togithub.com/googleapis/python-test-utils/blob/HEAD/CHANGELOG.md#​130-httpswwwgithubcomgoogleapispython-test-utilscomparev120v130-2021-11-16) [Compare Source](https://togithub.com/googleapis/python-test-utils/compare/v1.2.0...v1.3.0) ##### Features - add 'py.typed' declaration ([#​73](https://www.togithub.com/googleapis/python-test-utils/issues/73)) ([f8f5f0a](https://www.github.com/googleapis/python-test-utils/commit/f8f5f0a194b2420b2fee1cf88ac50220d3ba1538)) ### [`v1.2.0`](https://togithub.com/googleapis/python-test-utils/blob/HEAD/CHANGELOG.md#​120-httpswwwgithubcomgoogleapispython-test-utilscomparev110v120-2021-10-18) [Compare Source](https://togithub.com/googleapis/python-test-utils/compare/v1.1.0...v1.2.0) ##### Features - add support for python 3.10 ([#​68](https://www.togithub.com/googleapis/python-test-utils/issues/68)) ([d93b6a1](https://www.github.com/googleapis/python-test-utils/commit/d93b6a11e3bfade2b29ab90ed3bc2c384beb01cd)) </details> <details> <summary>pandas-dev/pandas</summary> ### [`v1.3.5`](https://togithub.com/pandas-dev/pandas/releases/v1.3.5) [Compare Source](https://togithub.com/pandas-dev/pandas/compare/v1.3.4...v1.3.5) This is a patch release in the 1.3.x series and includes some regression fixes. We recommend that all users upgrade to this version. See the [full whatsnew](https://pandas.pydata.org/pandas-docs/version/1.3.5/whatsnew/v1.3.5.html) for a list of all the changes. The release will be available on the defaults and conda-forge channels: conda install pandas Or via PyPI: python3 -m pip install --upgrade pandas Please report any issues with the release on the [pandas issue tracker](https://togithub.com/pandas-dev/pandas/issues). </details> --- ### Configuration 📅 **Schedule**: At any time (no schedule defined). 🚦 **Automerge**: Disabled by config. Please merge this manually once you are satisfied. ♻ **Rebasing**: Renovate will not automatically rebase this PR, because other commits have been found. 👻 **Immortal**: This PR will be recreated if closed unmerged. Get [config help](https://togithub.com/renovatebot/renovate/discussions) if that's undesired. --- - [ ] <!-- rebase-check -->If you want to rebase/retry this PR, click this checkbox. --- This PR has been generated by [WhiteSource Renovate](https://renovate.whitesourcesoftware.com). View repository job log [here](https://app.renovatebot.com/dashboard#github/googleapis/python-bigquery-pandas).
* chore(deps): update all dependencies * See googleapis/synthtool#1910 * See googleapis/synthtool#1910 --------- Co-authored-by: Anthonios Partheniou <partheniou@google.com>
* chore(deps): update actions/setup-python action to v4 * 🦉 Updates from OwlBot post-processor See https://github.com/googleapis/repo-automation-bots/blob/main/packages/owl-bot/README.md Co-authored-by: Anthonios Partheniou <partheniou@google.com> Co-authored-by: Owl Bot <gcf-owl-bot[bot]@users.noreply.github.com>
…ic enums (#419) * feat: enable "rest" transport in Python for services supporting numeric enums PiperOrigin-RevId: 508143576 Source-Link: googleapis/googleapis@7a702a9 Source-Link: googleapis/googleapis-gen@6ad1279 Copy-Tag: eyJwIjoiLmdpdGh1Yi8uT3dsQm90LnlhbWwiLCJoIjoiNmFkMTI3OWMwZTdhYTc4N2FjNmI2NmM5ZmQ0YTIxMDY5MmVkZmZjZCJ9 * 🦉 Updates from OwlBot post-processor See https://github.com/googleapis/repo-automation-bots/blob/main/packages/owl-bot/README.md * feat: Add `OR` query support docs: Minor documentation formatting and cleanup PiperOrigin-RevId: 509848726 Source-Link: googleapis/googleapis@b9351b4 Source-Link: googleapis/googleapis-gen@99317db Copy-Tag: eyJwIjoiLmdpdGh1Yi8uT3dsQm90LnlhbWwiLCJoIjoiOTkzMTdkYjY4MjcxODVlOGNhZThlZWI0NjEzN2NkODFjZDJkNmE1NCJ9 * 🦉 Updates from OwlBot post-processor See https://github.com/googleapis/repo-automation-bots/blob/main/packages/owl-bot/README.md * fix: Add service_yaml_parameters to py_gapic_library BUILD.bazel targets PiperOrigin-RevId: 510187992 Source-Link: googleapis/googleapis@5edc235 Source-Link: googleapis/googleapis-gen@b0bedb7 Copy-Tag: eyJwIjoiLmdpdGh1Yi8uT3dsQm90LnlhbWwiLCJoIjoiYjBiZWRiNzJlNDc2NWEzZTBiNjc0YTI4YzUwZWEwZjlhOWIyNmE4OSJ9 * 🦉 Updates from OwlBot post-processor See https://github.com/googleapis/repo-automation-bots/blob/main/packages/owl-bot/README.md * chore: Update gapic-generator-python to v1.8.5 PiperOrigin-RevId: 511892190 Source-Link: googleapis/googleapis@a45d9c0 Source-Link: googleapis/googleapis-gen@1907294 Copy-Tag: eyJwIjoiLmdpdGh1Yi8uT3dsQm90LnlhbWwiLCJoIjoiMTkwNzI5NGIxZDgzNjVlYTI0ZjhjNWYyZTA1OWE2NDEyNGM0ZWQzYiJ9 * 🦉 Updates from OwlBot post-processor See https://github.com/googleapis/repo-automation-bots/blob/main/packages/owl-bot/README.md --------- Co-authored-by: Owl Bot <gcf-owl-bot[bot]@users.noreply.github.com> Co-authored-by: Mariatta Wijaya <Mariatta@users.noreply.github.com> Co-authored-by: Anthonios Partheniou <partheniou@google.com>
Source-Link: googleapis/synthtool@5f2a608 Post-Processor: gcr.io/cloud-devrel-public-resources/owlbot-python:latest@sha256:8555f0e37e6261408f792bfd6635102d2da5ad73f8f09bcb24f25e6afb5fac97 Co-authored-by: Owl Bot <gcf-owl-bot[bot]@users.noreply.github.com>
Source-Link: googleapis/synthtool@7e1f6da Post-Processor: gcr.io/repo-automation-bots/owlbot-python:latest@sha256:a1a891041baa4ffbe1a809ac1b8b9b4a71887293c9101c88e8e255943c5aec2d Co-authored-by: Owl Bot <gcf-owl-bot[bot]@users.noreply.github.com>
…419) Co-authored-by: Anthonios Partheniou <partheniou@google.com>
…419) Co-authored-by: Anthonios Partheniou <partheniou@google.com>
* Use collections.abc, fixes #418 * Python 2.7 compat fix * Fix check
* Renaming storage gcloud samples folder. (#418)
* Add gcloud-based storage usage samples. (#419)
* Refactor cloud client storage samples. (#421)
* Add more storage samples for the cloud client libraries. (#432)
* Auto-update dependencies. (#456)
* Fix import order lint errors
Change-Id: Ieaf7237fc6f925daec46a07d2e81a452b841198a
* Add storage acl samples
Change-Id: Ib44f9bb42bf0c0607e64905a26369f06ea5fb231
* Address review comments
Change-Id: I94973a839f38ef3d1ec657c3c79f666eca56728b
* Fix lint issue
Change-Id: Ie9cf585303931f200a763d691906ad56221105fd
* Auto-update dependencies. (#540)
* Auto-update dependencies. (#542)
* Move to google-cloud (#544)
* Add new "quickstart" samples (#547)
* Quickstart tests (#569)
* Add tests for quickstarts
* Update secrets
* Add basic readme generator (#580)
* Generate readmes for most service samples (#599)
* Update samples to support latest Google Cloud Python (#656)
* Edited upload/download to perform encryption properly (#667)
* Storage Encryption Key Rotation Sample using Veneer + Tests (#672)
* Auto-update dependencies. (#715)
* Adds storage Pub/Sub notification polling tutorial (#875)
* Remove cloud config fixture (#887)
* Remove cloud config fixture
* Fix client secrets
* Fix bigtable instance
* Auto-update dependencies. (#914)
* Auto-update dependencies.
* xfail the error reporting test
* Fix lint
* Re-generate all readmes
* Add bucket-level IAM samples (#919)
* Add bucket-level IAM samples
* Address review comments
* Auto-update dependencies. (#927)
* Fix README rst links (#962)
* Fix README rst links
* Update all READMEs
* Auto-update dependencies. (#1004)
* Auto-update dependencies.
* Fix natural language samples
* Fix pubsub iam samples
* Fix language samples
* Fix bigquery samples
* Add bucket label samples (#1045)
* Auto-update dependencies. (#1055)
* Auto-update dependencies.
* Explicitly use latest bigtable client
Change-Id: Id71e9e768f020730e4ca9514a0d7ebaa794e7d9e
* Revert language update for now
Change-Id: I8867f154e9a5aae00d0047c9caf880e5e8f50c53
* Remove pdb. smh
Change-Id: I5ff905fadc026eebbcd45512d4e76e003e3b2b43
* Auto-update dependencies. (#1057)
* Auto-update dependencies. (#1073)
* Auto-update dependencies. (#1093)
* Auto-update dependencies.
* Fix storage notification poll sample
Change-Id: I6afbc79d15e050531555e4c8e51066996717a0f3
* Fix spanner samples
Change-Id: I40069222c60d57e8f3d3878167591af9130895cb
* Drop coverage because it's not useful
Change-Id: Iae399a7083d7866c3c7b9162d0de244fbff8b522
* Try again to fix flaky logging test
Change-Id: I6225c074701970c17c426677ef1935bb6d7e36b4
* Auto-update dependencies. (#1097)
* Update all generated readme auth instructions (#1121)
Change-Id: I03b5eaef8b17ac3dc3c0339fd2c7447bd3e11bd2
* Fix TypeError when running Storage notification polling exmaple. (#1135)
* Adds storage Pub/Sub notification polling tutorial
* Fix formatting and add some tests
* Auto-generate README
* Simplify implementation, remove classes
* Simplified example, removed de-duping
* regenerate README
* Remove explicit project parameter.
* Fix notification TypeError on start.
* Fix linter error.
* Fix ordered list ordinals.
* Rerun nox readmegen.
* Add support for overwrite attributes (#1142)
* Add support for overwrite attributes, bug fixes
* Lint fix for overwrite line
* Switch variable to snake_case
* Handle case where attribute not set (#1143)
* Added Link to Python Setup Guide (#1158)
* Update Readme.rst to add Python setup guide
As requested in b/64770713.
This sample is linked in documentation https://cloud.google.com/bigtable/docs/scaling, and it would make more sense to update the guide here than in the documentation.
* Update README.rst
* Update README.rst
* Update README.rst
* Update README.rst
* Update README.rst
* Update install_deps.tmpl.rst
* Updated readmegen scripts and re-generated related README files
* Fixed the lint error
* Auto-update dependencies. (#1138)
* storage requester pays samples (#1122)
* storage requester pays samples
* Added tests and fixed linting issues
* google-cloud-storage version update
* changed get_bucket to bucket for downloading
* small change
* Auto-update dependencies. (#1186)
* Auto-update dependencies. (#1234)
* Auto-update dependencies.
* Drop pytest-logcapture as it's no longer needed
Change-Id: Ia8b9e8aaf248e9770db6bc4842a4532df8383893
* Auto-update dependencies. (#1239)
* Added "Open in Cloud Shell" buttons to README files (#1254)
* Auto-update dependencies. (#1263)
* Auto-update dependencies. (#1272)
* Auto-update dependencies.
* Update requirements.txt
* Auto-update dependencies. (#1282)
* Auto-update dependencies.
* Fix storage acl sample
Change-Id: I413bea899fdde4c4859e4070a9da25845b81f7cf
* Auto-update dependencies. (#1320)
* Auto-update dependencies. (#1359)
* Auto-update dependencies. (#1377)
* Auto-update dependencies.
* Update requirements.txt
* Auto-update dependencies. (#1389)
* Auto-update dependencies.
* Regenerate the README files and fix the Open in Cloud Shell link for some samples (#1441)
* Update READMEs to fix numbering and add git clone (#1464)
* Fix typo. (#1509)
Fixes https://github.com/GoogleCloudPlatform/python-docs-samples/issues/1485
* Storage: add KMS samples (#1510)
* Storage: add KMS samples
* Add CLOUD_KMS_KEY environment variable
* [Storage] Update kms samples (#1517)
* Storage: add KMS samples
* Add CLOUD_KMS_KEY environment variable
* Add region tags around samples
* Add more testing
* Fix tests and lint
* Remove leftover merge conflict. (#1657)
* Add region tag to upload_blob snippet (#1671)
* Bucket lock samples (#1588)
* [Storage] Add spacing in sample code. (#1735)
* Add spacing in sample code.
* remove whitespace
* Auto-update dependencies. (#1846)
ACK, merging.
* Update requirements.txt (#1944)
* Update requirements.txt
* Adding some rate limiting
* Auto-update dependencies. (#1980)
* Auto-update dependencies.
* Update requirements.txt
* Update requirements.txt
* storage: bucket policy only samples (#1976)
* humble beginnings
* Verified integration tests and updated README.rst
* Updating samples to reflect fixed surface
* Use release 1.14.0
* Add sleep to avoid bucket rate limit (#2136)
* feat(storage): Add snippets for v4 signed URLs (#2142)
* feat(storage): Add snippets for v4 signed URLs
* lint
* fix .format()
* add v4 command to switch statement
* fix region tag
* change if => elif to try to make func less complex
* move main to a function
* storage: add list buckets (#2149)
* Add list_buckets sample
* Allow for more if conditions
* Drop xfail for passing test_remove_bucket_label (#2173)
The Python client was fixed in
https://github.com/googleapis/google-cloud-python/issues/3711
so the test now passes.
* Update string reported in snippet and update test
* Update list blobs to use new client.list_blobs() method. (#2296)
* Update list blob samples
* Update requirements.txt
* Fix lint issues
* Use latest storage client
* [Storage] Add comment to clarify which package version is necessary (#2315)
* Add comment to clarify which package version
* Lint and add another comment to related sample
* Storage: HMAC key samples (#2372)
Add samples for HMAC key functionality: list, create, get, activate, deactivate, delete. Includes tests and version bump for client library.
* Remove required argument from list buckets sample (#2394)
* Remove required argument from list buckets sample
* Remove required argument from list buckets sample
* Fixup sample for list_hmac_keys (#2404)
Correct printed metadata to match canonical samples.
* Bucket metadata sample (#2414)
* Remove required argument from list buckets sample
* Bucket metadata sample
* Bucket metadata sample
* Adds updates for samples profiler ... vision (#2439)
* fix: add bucket-name as required arg to v4 snippets (#2502)
* [Storage] Support rename of BPO to UniformBucketLevelAccess (#2335)
* Update BPO -> UBLA
* Update BPO -> UBLA
* Fix region tag (#2515)
* Update documentation for prefix, delimiter search (#2537)
* Update documentation for prefix, delimiter search
* Remove whitespace.
* [Storage] Split samples (#2602)
* split bucket_lock samples and lint
* split samples
* blacken
* fix typos
* Add missing tests and lint
* lint
* fix typos
* fix typo
* typo
* remove README
* Auto-update dependencies. (#2005)
* Auto-update dependencies.
* Revert update of appengine/flexible/datastore.
* revert update of appengine/flexible/scipy
* revert update of bigquery/bqml
* revert update of bigquery/cloud-client
* revert update of bigquery/datalab-migration
* revert update of bigtable/quickstart
* revert update of compute/api
* revert update of container_registry/container_analysis
* revert update of dataflow/run_template
* revert update of datastore/cloud-ndb
* revert update of dialogflow/cloud-client
* revert update of dlp
* revert update of functions/imagemagick
* revert update of functions/ocr/app
* revert update of healthcare/api-client/fhir
* revert update of iam/api-client
* revert update of iot/api-client/gcs_file_to_device
* revert update of iot/api-client/mqtt_example
* revert update of language/automl
* revert update of run/image-processing
* revert update of vision/automl
* revert update testing/requirements.txt
* revert update of vision/cloud-client/detect
* revert update of vision/cloud-client/product_search
* revert update of jobs/v2/api_client
* revert update of jobs/v3/api_client
* revert update of opencensus
* revert update of translate/cloud-client
* revert update to speech/cloud-client
Co-authored-by: Kurtis Van Gent <31518063+kurtisvg@users.noreply.github.com>
Co-authored-by: Doug Mahugh <dmahugh@gmail.com>
* samples(storage): IAM conditions samples (#2730)
* docs(storage): use policy.bindings in Storage/IAM samples
* update view Bucket IAM to use policy.bindings
* update remove Bucket IAM to use policy.bindings
* blacken
* add IAM condition sample
* add conditional iam binding sample
* bump storage requirement to 1.25.0
* fix tests
* remove unused imports
* fix: Use unique resources for storage snippets. (#3029)
* fix: use unique buckets and blobs for acl tests
* fix: use unique buckets and blobs for snippets tests
* fix: reuse test_bucket within module to avoid exhausting quota
* fix: Due to retention policy, don't reuse fixture for bucket lock
* fix: randomize blob names to disperse file edits
* fix: Reuse HMAC key as we have a limit of 5 (#3037)
* fix: Reuse HMAC key as we have a limit of 5
* fix: harden storage test fixtures (#3039)
* fix: improve UBLA test fixtures
* fix: improve IAM test fixtures
* storage: Fix docs for signed URL generation (#3008)
Co-authored-by: Bu Sun Kim <8822365+busunkim96@users.noreply.github.com>
Co-authored-by: Christopher Wilcox <crwilcox@google.com>
* chore(deps): update dependency google-cloud-storage to v1.26.0 (#3046)
* chore(deps): update dependency google-cloud-storage to v1.26.0
* chore(deps): specify dependencies by python version
* chore: up other deps to try to remove errors
Co-authored-by: Leah E. Cole <6719667+leahecole@users.noreply.github.com>
Co-authored-by: Leah Cole <coleleah@google.com>
* feat: add remove conditional binding sample (#3107)
* feat: add remove conditional binding sample
* fix iam test fixture
* fix silly mistake of removing all bindings
* fix ubla test
* address feedback
* revert changes to tests
* Simplify noxfile setup. (#2806)
* chore(deps): update dependency requests to v2.23.0
* Simplify noxfile and add version control.
* Configure appengine/standard to only test Python 2.7.
* Update Kokokro configs to match noxfile.
* Add requirements-test to each folder.
* Remove Py2 versions from everything execept appengine/standard.
* Remove conftest.py.
* Remove appengine/standard/conftest.py
* Remove 'no-sucess-flaky-report' from pytest.ini.
* Add GAE SDK back to appengine/standard tests.
* Fix typo.
* Roll pytest to python 2 version.
* Add a bunch of testing requirements.
* Remove typo.
* Add appengine lib directory back in.
* Add some additional requirements.
* Fix issue with flake8 args.
* Even more requirements.
* Readd appengine conftest.py.
* Add a few more requirements.
* Even more Appengine requirements.
* Add webtest for appengine/standard/mailgun.
* Add some additional requirements.
* Add workaround for issue with mailjet-rest.
* Add responses for appengine/standard/mailjet.
Co-authored-by: Renovate Bot <bot@renovateapp.com>
* [storage] feat: add post policy sample (#3231)
* feat: add post policy sample
* use 1.27.0
* fix
* simplify iterator
Co-authored-by: Jonathan Lui <jonathanlui@google.com>
* Update dependency google-cloud-pubsub to v1.4.2 in Storage and Pub/Sub (#3343)
* Remove name attribute from the input (#3569)
If name='submit' is specified for the input type='submit' the endpoint returns the following error:
<Error>
<Code>InvalidPolicyDocument</Code>
<Message>
The content of the form does not meet the conditions specified in the policy document.
</Message>
<Details>Policy did not reference these fields: submit</Details>
</Error>
Co-authored-by: Takashi Matsuo <tmatsuo@google.com>
* [storage] fix: use unique blob name (#3568)
* [storage] fix: use unique blob name
fixes #3567
* add some comments
* chore(deps): update dependency google-cloud-storage to v1.28.0 (#3260)
Co-authored-by: Takashi Matsuo <tmatsuo@google.com>
* [storage] fix: use a different bucket for requester_pays_test (#3655)
* [storage] fix: use a different bucket for requester_pays_test
fixes #3654
* rename to README.md, added the envvar to the template
* add REQUESTER_PAYS_TEST_BUCKET env var
* just use REQUESTER_PAYS_TEST_BUCKET
* docs(storage): add samples for lifer cycle and versioning (#3578)
* docs(storage): add samples for lifer cycle and versioning
* docs(storage): nits
* docs(storage): lint fix
Co-authored-by: Leah E. Cole <6719667+leahecole@users.noreply.github.com>
* chore: some lint fixes (#3750)
* chore(deps): update dependency google-cloud-pubsub to v1.4.3 (#3725)
Co-authored-by: Bu Sun Kim <8822365+busunkim96@users.noreply.github.com>
Co-authored-by: Takashi Matsuo <tmatsuo@google.com>
* docs(storage): add samples (#3687)
* chore(deps): update dependency google-cloud-storage to v1.28.1 (#3785)
* chore(deps): update dependency google-cloud-storage to v1.28.1
* [asset] testing: use uuid instead of time
Co-authored-by: Takashi Matsuo <tmatsuo@google.com>
* docs(storage): add samples for file archive generation and cors configuration (#3794)
* chore(deps): update dependency google-cloud-pubsub to v1.5.0 (#3781)
Co-authored-by: Bu Sun Kim <8822365+busunkim96@users.noreply.github.com>
* Replace GCLOUD_PROJECT with GOOGLE_CLOUD_PROJECT. (#4022)
* [storage] testing: use multiple projects (#4048)
* [storage] testing: use multiple projects
We still need to use the old project for some tests.
fixes #4033
fixes #4029
* remove print
* use uuid instead of time.time()
* lint fix
* chore(deps): update dependency google-cloud-storage to v1.29.0 (#4040)
* Update dependency google-cloud-pubsub to v1.6.0 (#4039)
This PR contains the following updates:
| Package | Update | Change |
|---|---|---|
| [google-cloud-pubsub](https://togithub.com/googleapis/python-pubsub) | minor | `==1.5.0` -> `==1.6.0` |
---
### Release Notes
<details>
<summary>googleapis/python-pubsub</summary>
### [`v1.6.0`](https://togithub.com/googleapis/python-pubsub/blob/master/CHANGELOG.md#​160-httpswwwgithubcomgoogleapispython-pubsubcomparev150v160-2020-06-09)
[Compare Source](https://togithub.com/googleapis/python-pubsub/compare/v1.5.0...v1.6.0)
##### Features
- Add flow control for message publishing ([#​96](https://www.github.com/googleapis/python-pubsub/issues/96)) ([06085c4](https://www.github.com/googleapis/python-pubsub/commit/06085c4083b9dccdd50383257799904510bbf3a0))
##### Bug Fixes
- Fix PubSub incompatibility with api-core 1.17.0+ ([#​103](https://www.github.com/googleapis/python-pubsub/issues/103)) ([c02060f](https://www.github.com/googleapis/python-pubsub/commit/c02060fbbe6e2ca4664bee08d2de10665d41dc0b))
##### Documentation
- Clarify that Schedulers shouldn't be used with multiple SubscriberClients ([#​100](https://togithub.com/googleapis/python-pubsub/pull/100)) ([cf9e87c](https://togithub.com/googleapis/python-pubsub/commit/cf9e87c80c0771f3fa6ef784a8d76cb760ad37ef))
- Fix update subscription/snapshot/topic samples ([#​113](https://togithub.com/googleapis/python-pubsub/pull/113)) ([e62c38b](https://togithub.com/googleapis/python-pubsub/commit/e62c38bb33de2434e32f866979de769382dea34a))
##### Internal / Testing Changes
- Re-generated service implementaton using synth: removed experimental notes from the RetryPolicy and filtering features in anticipation of GA, added DetachSubscription (experimental) ([#​114](https://togithub.com/googleapis/python-pubsub/pull/114)) ([0132a46](https://togithub.com/googleapis/python-pubsub/commit/0132a4680e0727ce45d5e27d98ffc9f3541a0962))
- Incorporate will_accept() checks into publish() ([#​108](https://togithub.com/googleapis/python-pubsub/pull/108)) ([6c7677e](https://togithub.com/googleapis/python-pubsub/commit/6c7677ecb259672bbb9b6f7646919e602c698570))
</details>
---
### Renovate configuration
:date: **Schedule**: At any time (no schedule defined).
:vertical_traffic_light: **Automerge**: Disabled by config. Please merge this manually once you are satisfied.
:recycle: **Rebasing**: Never, or you tick the rebase/retry checkbox.
:no_bell: **Ignore**: Close this PR and you won't be reminded about this update again.
---
- [ ] <!-- rebase-check -->If you want to rebase/retry this PR, check this box
---
This PR has been generated by [WhiteSource Renovate](https://renovate.whitesourcesoftware.com). View repository job log [here](https://app.renovatebot.com/dashboard#GoogleCloudPlatform/python-docs-samples).
* chore(deps): update dependency google-cloud-pubsub to v1.6.1 (#4242)
Co-authored-by: gcf-merge-on-green[bot] <60162190+gcf-merge-on-green[bot]@users.noreply.github.com>
* chore(deps): update dependency pytest to v5.4.3 (#4279)
* chore(deps): update dependency pytest to v5.4.3
* specify pytest for python 2 in appengine
Co-authored-by: Leah Cole <coleleah@google.com>
* chore(deps): update dependency mock to v4 (#4287)
* chore(deps): update dependency mock to v4
* specify mock version for appengine python 2
Co-authored-by: Leah Cole <coleleah@google.com>
* chore(deps): update dependency google-cloud-pubsub to v1.7.0 (#4290)
This PR contains the following updates:
| Package | Update | Change |
|---|---|---|
| [google-cloud-pubsub](https://togithub.com/googleapis/python-pubsub) | minor | `==1.6.1` -> `==1.7.0` |
---
### Release Notes
<details>
<summary>googleapis/python-pubsub</summary>
### [`v1.7.0`](https://togithub.com/googleapis/python-pubsub/blob/master/CHANGELOG.md#​170-httpswwwgithubcomgoogleapispython-pubsubcomparev161v170-2020-07-13)
[Compare Source](https://togithub.com/googleapis/python-pubsub/compare/v1.6.1...v1.7.0)
##### New Features
- Add support for server-side flow control. ([#​143](https://togithub.com/googleapis/python-pubsub/pull/143)) ([04e261c](https://www.github.com/googleapis/python-pubsub/commit/04e261c602a2919cc75b3efa3dab099fb2cf704c))
##### Dependencies
- Update samples dependency `google-cloud-pubsub` to `v1.6.1`. ([#​144](https://togithub.com/googleapis/python-pubsub/pull/144)) ([1cb6746](https://togithub.com/googleapis/python-pubsub/commit/1cb6746b00ebb23dbf1663bae301b32c3fc65a88))
##### Documentation
- Add pubsub/cloud-client samples from the common samples repo (with commit history). ([#​151](https://togithub.com/googleapis/python-pubsub/pull/151))
- Add flow control section to publish overview. ([#​129](https://togithub.com/googleapis/python-pubsub/pull/129)) ([acc19eb](https://www.github.com/googleapis/python-pubsub/commit/acc19eb048eef067d9818ef3e310b165d9c6307e))
- Add a link to Pub/Sub filtering language public documentation to `pubsub.proto`. ([#​121](https://togithub.com/googleapis/python-pubsub/pull/121)) ([8802d81](https://www.github.com/googleapis/python-pubsub/commit/8802d8126247f22e26057e68a42f5b5a82dcbf0d))
</details>
---
### Renovate configuration
:date: **Schedule**: At any time (no schedule defined).
:vertical_traffic_light: **Automerge**: Disabled by config. Please merge this manually once you are satisfied.
:recycle: **Rebasing**: Renovate will not automatically rebase this PR, because other commits have been found.
:no_bell: **Ignore**: Close this PR and you won't be reminded about this update again.
---
- [ ] <!-- rebase-check -->If you want to rebase/retry this PR, check this box
---
This PR has been generated by [WhiteSource Renovate](https://renovate.whitesourcesoftware.com). View repository job log [here](https://app.renovatebot.com/dashboard#GoogleCloudPlatform/python-docs-samples).
* Fix mismatched storage region tags (#4194)
* Update dependency google-cloud-storage to v1.30.0
* Update dependency pytest to v6 (#4390)
* chore(deps): update dependency google-cloud-storage to v1.31.0 (#4564)
Co-authored-by: Takashi Matsuo <tmatsuo@google.com>
* chore: fix some more unmatched region tags (#4585)
fixes #4549
Co-authored-by: Dina Graves Portman <dinagraves@google.com>
* Update storage_get_metadata.py (#4615)
Co-authored-by: Leah E. Cole <6719667+leahecole@users.noreply.github.com>
* chore(deps): update dependency google-cloud-storage to v1.31.1 (#4714)
* chore(deps): update dependency google-cloud-storage to v1.31.2 (#4750)
This PR contains the following updates:
| Package | Update | Change |
|---|---|---|
| [google-cloud-storage](https://togithub.com/googleapis/python-storage) | patch | `==1.31.1` -> `==1.31.2` |
---
### Release Notes
<details>
<summary>googleapis/python-storage</summary>
### [`v1.31.2`](https://togithub.com/googleapis/python-storage/blob/master/CHANGELOG.md#​1312-httpswwwgithubcomgoogleapispython-storagecomparev1311v1312-2020-09-23)
[Compare Source](https://togithub.com/googleapis/python-storage/compare/v1.31.1...v1.31.2)
</details>
---
### Renovate configuration
:date: **Schedule**: At any time (no schedule defined).
:vertical_traffic_light: **Automerge**: Disabled by config. Please merge this manually once you are satisfied.
:recycle: **Rebasing**: Never, or you tick the rebase/retry checkbox.
:no_bell: **Ignore**: Close this PR and you won't be reminded about this update again.
---
- [ ] <!-- rebase-check -->If you want to rebase/retry this PR, check this box
---
This PR has been generated by [WhiteSource Renovate](https://renovate.whitesourcesoftware.com). View repository job log [here](https://app.renovatebot.com/dashboard#github/GoogleCloudPlatform/python-docs-samples).
* chore(deps): update dependency pytest to v6.1.1 (#4761)
* chore(deps): update dependency google-cloud-storage to v1.32.0 (#4871)
* chore(deps): update dependency pytest to v6.1.2 (#4921)
Co-authored-by: Charles Engelke <engelke@google.com>
* change pprint to print. (#4856)
* change pprint to print.
Line 57 had pprint.pprint.. changing it to print.
* Update storage_get_bucket_metadata.py
Removing pprint import
Co-authored-by: Dina Graves Portman <dinagraves@google.com>
Co-authored-by: Charles Engelke <engelke@google.com>
* chore(deps): update dependency google-cloud-storage to v1.33.0 (#4990)
* Add patch call (#5013)
I believe a call to `blob.patch()` is necessary to actually save the metadata back to GCS.
* fix: add a comment to draw attention to using get_blob, not blob (#5052)
* fix: add a comment to draw attention to using get_blob, not blob
* docs: further elaboration
* docs: add clarifying doc string to download file
* Update storage_download_file.py
* Update storage_download_file.py
* chore(deps): update dependency mock to v4.0.3 (#5062)
* fix(storage): Update comment, prefix should include delimiter (#5064)
* chore(deps): update dependency google-cloud-storage to v1.35.0 (#5074)
* chore(deps): update dependency pytest to v6.2.1 (#5076)
[](https://renovatebot.com)
This PR contains the following updates:
| Package | Update | Change |
|---|---|---|
| [pytest](https://docs.pytest.org/en/latest/) ([source](https://togithub.com/pytest-dev/pytest)) | minor | `==6.1.2` -> `==6.2.1` |
---
### Release Notes
<details>
<summary>pytest-dev/pytest</summary>
### [`v6.2.1`](https://togithub.com/pytest-dev/pytest/releases/6.2.1)
[Compare Source](https://togithub.com/pytest-dev/pytest/compare/6.2.0...6.2.1)
# pytest 6.2.1 (2020-12-15)
## Bug Fixes
- [#​7678](https://togithub.com/pytest-dev/pytest/issues/7678): Fixed bug where `ImportPathMismatchError` would be raised for files compiled in
the host and loaded later from an UNC mounted path (Windows).
- [#​8132](https://togithub.com/pytest-dev/pytest/issues/8132): Fixed regression in `approx`: in 6.2.0 `approx` no longer raises
`TypeError` when dealing with non-numeric types, falling back to normal comparison.
Before 6.2.0, array types like tf.DeviceArray fell through to the scalar case,
and happened to compare correctly to a scalar if they had only one element.
After 6.2.0, these types began failing, because they inherited neither from
standard Python number hierarchy nor from `numpy.ndarray`.
`approx` now converts arguments to `numpy.ndarray` if they expose the array
protocol and are not scalars. This treats array-like objects like numpy arrays,
regardless of size.
### [`v6.2.0`](https://togithub.com/pytest-dev/pytest/releases/6.2.0)
[Compare Source](https://togithub.com/pytest-dev/pytest/compare/6.1.2...6.2.0)
# pytest 6.2.0 (2020-12-12)
## Breaking Changes
- [#​7808](https://togithub.com/pytest-dev/pytest/issues/7808): pytest now supports python3.6+ only.
## Deprecations
- [#​7469](https://togithub.com/pytest-dev/pytest/issues/7469): Directly constructing/calling the following classes/functions is now deprecated:
- `_pytest.cacheprovider.Cache`
- `_pytest.cacheprovider.Cache.for_config()`
- `_pytest.cacheprovider.Cache.clear_cache()`
- `_pytest.cacheprovider.Cache.cache_dir_from_config()`
- `_pytest.capture.CaptureFixture`
- `_pytest.fixtures.FixtureRequest`
- `_pytest.fixtures.SubRequest`
- `_pytest.logging.LogCaptureFixture`
- `_pytest.pytester.Pytester`
- `_pytest.pytester.Testdir`
- `_pytest.recwarn.WarningsRecorder`
- `_pytest.recwarn.WarningsChecker`
- `_pytest.tmpdir.TempPathFactory`
- `_pytest.tmpdir.TempdirFactory`
These have always been considered private, but now issue a deprecation warning, which may become a hard error in pytest 7.0.0.
- [#​7530](https://togithub.com/pytest-dev/pytest/issues/7530): The `--strict` command-line option has been deprecated, use `--strict-markers` instead.
We have plans to maybe in the future to reintroduce `--strict` and make it an encompassing flag for all strictness
related options (`--strict-markers` and `--strict-config` at the moment, more might be introduced in the future).
- [#​7988](https://togithub.com/pytest-dev/pytest/issues/7988): The `@pytest.yield_fixture` decorator/function is now deprecated. Use pytest.fixture instead.
`yield_fixture` has been an alias for `fixture` for a very long time, so can be search/replaced safely.
## Features
- [#​5299](https://togithub.com/pytest-dev/pytest/issues/5299): pytest now warns about unraisable exceptions and unhandled thread exceptions that occur in tests on Python>=3.8.
See unraisable for more information.
- [#​7425](https://togithub.com/pytest-dev/pytest/issues/7425): New pytester fixture, which is identical to testdir but its methods return pathlib.Path when appropriate instead of `py.path.local`.
This is part of the movement to use pathlib.Path objects internally, in order to remove the dependency to `py` in the future.
Internally, the old Testdir <\_pytest.pytester.Testdir> is now a thin wrapper around Pytester <\_pytest.pytester.Pytester>, preserving the old interface.
- [#​7695](https://togithub.com/pytest-dev/pytest/issues/7695): A new hook was added, pytest_markeval_namespace which should return a dictionary.
This dictionary will be used to augment the "global" variables available to evaluate skipif/xfail/xpass markers.
Pseudo example
`conftest.py`:
```{.sourceCode .python}
def pytest_markeval_namespace():
return {"color": "red"}
```
`test_func.py`:
```{.sourceCode .python}
@​pytest.mark.skipif("color == 'blue'", reason="Color is not red")
def test_func():
assert False
```
- [#​8006](https://togithub.com/pytest-dev/pytest/issues/8006): It is now possible to construct a ~pytest.MonkeyPatch object directly as `pytest.MonkeyPatch()`,
in cases when the monkeypatch fixture cannot be used. Previously some users imported it
from the private \_pytest.monkeypatch.MonkeyPatch namespace.
Additionally, MonkeyPatch.context <pytest.MonkeyPatch.context> is now a classmethod,
and can be used as `with MonkeyPatch.context() as mp: ...`. This is the recommended way to use
`MonkeyPatch` directly, since unlike the `monkeypatch` fixture, an instance created directly
is not `undo()`-ed automatically.
## Improvements
- [#​1265](https://togithub.com/pytest-dev/pytest/issues/1265): Added an `__str__` implementation to the ~pytest.pytester.LineMatcher class which is returned from `pytester.run_pytest().stdout` and similar. It returns the entire output, like the existing `str()` method.
- [#​2044](https://togithub.com/pytest-dev/pytest/issues/2044): Verbose mode now shows the reason that a test was skipped in the test's terminal line after the "SKIPPED", "XFAIL" or "XPASS".
- [#​7469](https://togithub.com/pytest-dev/pytest/issues/7469) The types of builtin pytest fixtures are now exported so they may be used in type annotations of test functions.
The newly-exported types are:
- `pytest.FixtureRequest` for the request fixture.
- `pytest.Cache` for the cache fixture.
- `pytest.CaptureFixture[str]` for the capfd and capsys fixtures.
- `pytest.CaptureFixture[bytes]` for the capfdbinary and capsysbinary fixtures.
- `pytest.LogCaptureFixture` for the caplog fixture.
- `pytest.Pytester` for the pytester fixture.
- `pytest.Testdir` for the testdir fixture.
- `pytest.TempdirFactory` for the tmpdir_factory fixture.
- `pytest.TempPathFactory` for the tmp_path_factory fixture.
- `pytest.MonkeyPatch` for the monkeypatch fixture.
- `pytest.WarningsRecorder` for the recwarn fixture.
Constructing them is not supported (except for MonkeyPatch); they are only meant for use in type annotations.
Doing so will emit a deprecation warning, and may become a hard-error in pytest 7.0.
Subclassing them is also not supported. This is not currently enforced at runtime, but is detected by type-checkers such as mypy.
- [#​7527](https://togithub.com/pytest-dev/pytest/issues/7527): When a comparison between namedtuple <collections.namedtuple> instances of the same type fails, pytest now shows the differing field names (possibly nested) instead of their indexes.
- [#​7615](https://togithub.com/pytest-dev/pytest/issues/7615): Node.warn <\_pytest.nodes.Node.warn> now permits any subclass of Warning, not just PytestWarning <pytest.PytestWarning>.
- [#​7701](https://togithub.com/pytest-dev/pytest/issues/7701): Improved reporting when using `--collected-only`. It will now show the number of collected tests in the summary stats.
- [#​7710](https://togithub.com/pytest-dev/pytest/issues/7710): Use strict equality comparison for non-numeric types in pytest.approx instead of
raising TypeError.
This was the undocumented behavior before 3.7, but is now officially a supported feature.
- [#​7938](https://togithub.com/pytest-dev/pytest/issues/7938): New `--sw-skip` argument which is a shorthand for `--stepwise-skip`.
- [#​8023](https://togithub.com/pytest-dev/pytest/issues/8023): Added `'node_modules'` to default value for norecursedirs.
- [#​8032](https://togithub.com/pytest-dev/pytest/issues/8032): doClassCleanups <unittest.TestCase.doClassCleanups> (introduced in unittest in Python and 3.8) is now called appropriately.
## Bug Fixes
- [#​4824](https://togithub.com/pytest-dev/pytest/issues/4824): Fixed quadratic behavior and improved performance of collection of items using autouse fixtures and xunit fixtures.
- [#​7758](https://togithub.com/pytest-dev/pytest/issues/7758): Fixed an issue where some files in packages are getting lost from `--lf` even though they contain tests that failed. Regressed in pytest 5.4.0.
- [#​7911](https://togithub.com/pytest-dev/pytest/issues/7911): Directories created by by tmp_path and tmpdir are now considered stale after 3 days without modification (previous value was 3 hours) to avoid deleting directories still in use in long running test suites.
- [#​7913](https://togithub.com/pytest-dev/pytest/issues/7913): Fixed a crash or hang in pytester.spawn <\_pytest.pytester.Pytester.spawn> when the readline module is involved.
- [#​7951](https://togithub.com/pytest-dev/pytest/issues/7951): Fixed handling of recursive symlinks when collecting tests.
- [#​7981](https://togithub.com/pytest-dev/pytest/issues/7981): Fixed symlinked directories not being followed during collection. Regressed in pytest 6.1.0.
- [#​8016](https://togithub.com/pytest-dev/pytest/issues/8016): Fixed only one doctest being collected when using `pytest --doctest-modules path/to/an/__init__.py`.
## Improved Documentation
- [#​7429](https://togithub.com/pytest-dev/pytest/issues/7429): Add more information and use cases about skipping doctests.
- [#​7780](https://togithub.com/pytest-dev/pytest/issues/7780): Classes which should not be inherited from are now marked `final class` in the API reference.
- [#​7872](https://togithub.com/pytest-dev/pytest/issues/7872): `_pytest.config.argparsing.Parser.addini()` accepts explicit `None` and `"string"`.
- [#​7878](https://togithub.com/pytest-dev/pytest/issues/7878): In pull request section, ask to commit after editing changelog and authors file.
## Trivial/Internal Changes
- [#​7802](https://togithub.com/pytest-dev/pytest/issues/7802): The `attrs` dependency requirement is now >=19.2.0 instead of >=17.4.0.
- [#​8014](https://togithub.com/pytest-dev/pytest/issues/8014): .pyc files created by pytest's assertion rewriting now conform to the newer PEP-552 format on Python>=3.7.
(These files are internal and only interpreted by pytest itself.)
</details>
---
### Renovate configuration
:date: **Schedule**: At any time (no schedule defined).
:vertical_traffic_light: **Automerge**: Disabled by config. Please merge this manually once you are satisfied.
:recycle: **Rebasing**: Never, or you tick the rebase/retry checkbox.
:no_bell: **Ignore**: Close this PR and you won't be reminded about this update again.
---
- [ ] <!-- rebase-check -->If you want to rebase/retry this PR, check this box
---
This PR has been generated by [WhiteSource Renovate](https://renovate.whitesourcesoftware.com). View repository job log [here](https://app.renovatebot.com/dashboard#github/GoogleCloudPlatform/python-docs-samples).
* chore(deps): update dependency google-cloud-pubsub to v2.2.0 (#4673)
* chore(deps): update dependency google-cloud-pubsub to v2.2.0
* run pubsub script on healthcare/api-client/v1/dicom
* iot pubsub fixes, fix lint
* revert some changes pubsub script made
* try using return_value for mock
* undo previous change
* try adding publish_time in mock
* move publish_time param
* make publish_time a float
* make publish_time a datetime
* try using object instead of datetime
* another attempt
* undo the black stuff that messed up lint
Co-authored-by: Leah Cole <coleleah@google.com>
Co-authored-by: Leah E. Cole <6719667+leahecole@users.noreply.github.com>
* chore(Dockerfile): add Python 3.9 (#4968)
* chore(Dockerfile): add Python 3.9
* Add py3.9 kokoro dir
* fix typo
* Add GPG keys
* Add 3.9 to noxfiles
* Update composer dep to avoid deprecation spam
* fix(storage): add py-3.9 specific key
* update psycopg2-binary, only run test in py-3.9 build
* add libmemcached-dev to the Dockerfile
* disable appengine standard test in py-3.9 build
* disable py-3.9 build for appengine cloud_debugger
* skip py-3.9 build for composer/workflows
* skip tests with pyarrow for py-3.9 build
* avoid ReferenceError in iot builds
* skip some tests due to pip error
* add a temporary statement for debugging
* fix lint
* use correct constant
* disable 2.7 builds
* disable builds due to pip conflict
The conflict is between google-cloud-monitoring==2.0.0 and
opencensus-ext-stackdriver.
* remove temporary debugging statement
* really skip py-3.9 build for pubsub/streaming-analytics
* copyright year fix
* fix(storage): explicitly use the test project for the test bucket
* fix(storage): use correct cloud project
* fix: disable py-3.9 builds
- appengine/standard_python3/bigquery
- data-science-onramp/data-ingestion
* disable py-3.9 build
- dataflow/encryption-keys
- dataflow/flex-templates/streaming_beam
* disable type hint checks
Co-authored-by: Takashi Matsuo <tmatsuo@google.com>
* fix(storage): list all versions (#5325)
## Description
Add the `versions=True` variable to the `list_file_archived_generations function` to actually list all the versions instead of the last one only.
Fixes the incongruency between python and the other languages in the [Listing noncurrent object versions code samples](https://cloud.google.com/storage/docs/using-object-versioning#list).
## Checklist
- [x] I have followed [Sample Guidelines from AUTHORING_GUIDE.MD](https://github.com/GoogleCloudPlatform/python-docs-samples/blob/master/AUTHORING_GUIDE.md)
- [x] README is updated to include [all relevant information](https://github.com/GoogleCloudPlatform/python-docs-samples/blob/master/AUTHORING_GUIDE.md#readme-file)
- [x] **Tests** pass: `nox -s py-3.6` (see [Test Environment Setup](https://github.com/GoogleCloudPlatform/python-docs-samples/blob/master/AUTHORING_GUIDE.md#test-environment-setup))
- [x] **Lint** pass: `nox -s lint` (see [Test Environment Setup](https://github.com/GoogleCloudPlatform/python-docs-samples/blob/master/AUTHORING_GUIDE.md#test-environment-setup))
- [ ] These samples need a new **API enabled** in testing projects to pass (let us know which ones)
- [ ] These samples need a new/updated **env vars** in testing projects set to pass (let us know which ones)
- [x] Please **merge** this PR for me once it is approved.
- [ ] This sample adds a new sample directory, and I updated the [CODEOWNERS file](https://github.com/GoogleCloudPlatform/python-docs-samples/blob/master/.github/CODEOWNERS) with the codeowners for this sample
* docs: address sample feedback issues (#5329)
## Description
Fixes #5180, captures work from #5181 authored by @keegan2149, thank you!
## Checklist
- [x] I have followed [Sample Guidelines from AUTHORING_GUIDE.MD](https://github.com/GoogleCloudPlatform/python-docs-samples/blob/master/AUTHORING_GUIDE.md)
- [x] README is updated to include [all relevant information](https://github.com/GoogleCloudPlatform/python-docs-samples/blob/master/AUTHORING_GUIDE.md#readme-file)
- [x] **Tests** pass: `nox -s py-3.6` (see [Test Environment Setup](https://github.com/GoogleCloudPlatform/python-docs-samples/blob/master/AUTHORING_GUIDE.md#test-environment-setup))
- [x] **Lint** pass: `nox -s lint` (see [Test Environment Setup](https://github.com/GoogleCloudPlatform/python-docs-samples/blob/master/AUTHORING_GUIDE.md#test-environment-setup))
- [x] Please **merge** this PR for me once it is approved.
* chore(deps): update dependency google-cloud-pubsub to v2.3.0 (#5347)
[](https://renovatebot.com)
This PR contains the following updates:
| Package | Change | Age | Adoption | Passing | Confidence |
|---|---|---|---|---|---|
| [google-cloud-pubsub](https://togithub.com/googleapis/python-pubsub) | `==2.2.0` -> `==2.3.0` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) |
---
### Release Notes
<details>
<summary>googleapis/python-pubsub</summary>
### [`v2.3.0`](https://togithub.com/googleapis/python-pubsub/blob/master/CHANGELOG.md#​230-httpswwwgithubcomgoogleapispython-pubsubcomparev220v230-2021-02-08)
[Compare Source](https://togithub.com/googleapis/python-pubsub/compare/v2.2.0...v2.3.0)
##### Features
- surface SchemaServiceClient in google.cloud.pubsub ([#​281](https://www.github.com/googleapis/python-pubsub/issues/281)) ([8751bcc](https://www.github.com/googleapis/python-pubsub/commit/8751bcc5eb782df55769b48253629a3bde3d4661))
##### Bug Fixes
- client version missing from the user agent header ([#​275](https://www.github.com/googleapis/python-pubsub/issues/275)) ([b112f4f](https://www.github.com/googleapis/python-pubsub/commit/b112f4fcbf6f2bce8dcf37871bdc540b11f54fe3))
- Don't open the google.cloud package by adding pubsub.py ([#​269](https://www.github.com/googleapis/python-pubsub/issues/269)) ([542d79d](https://www.github.com/googleapis/python-pubsub/commit/542d79d7c5fb7403016150ba477485756cd4097b))
- flaky samples tests ([#​263](https://www.github.com/googleapis/python-pubsub/issues/263)) ([3d6a29d](https://www.github.com/googleapis/python-pubsub/commit/3d6a29de07cc09be663c90a3333f4cd33633994f))
- Modify synth.py to update grpc transport options ([#​266](https://www.github.com/googleapis/python-pubsub/issues/266)) ([41dcd30](https://www.github.com/googleapis/python-pubsub/commit/41dcd30636168f3dd1248f1d99170d531fc9bcb8))
- pass anonymous credentials for emulator ([#​250](https://www.github.com/googleapis/python-pubsub/issues/250)) ([8eed8e1](https://www.github.com/googleapis/python-pubsub/commit/8eed8e16019510dc8b20fb6b009d61a7ac532d26))
- remove grpc send/recieve limits ([#​259](https://www.github.com/googleapis/python-pubsub/issues/259)) ([fd2840c](https://www.github.com/googleapis/python-pubsub/commit/fd2840c10f92b03da7f4b40ac69c602220757c0a))
</details>
---
### Renovate configuration
:date: **Schedule**: At any time (no schedule defined).
:vertical_traffic_light: **Automerge**: Disabled by config. Please merge this manually once you are satisfied.
:recycle: **Rebasing**: Never, or you tick the rebase/retry checkbox.
:no_bell: **Ignore**: Close this PR and you won't be reminded about this update again.
---
- [ ] <!-- rebase-check -->If you want to rebase/retry this PR, check this box
---
This PR has been generated by [WhiteSource Renovate](https://renovate.whitesourcesoftware.com). View repository job log [here](https://app.renovatebot.com/dashboard#github/GoogleCloudPlatform/python-docs-samples).
* chore(deps): update dependency google-cloud-storage to v1.35.1 (#5321)
* chore(deps): update dependency google-cloud-pubsub to v2.4.0 (#5399)
* chore(deps): update dependency google-cloud-storage to v1.36.1 (#5353)
* chore(deps): update dependency google-cloud-storage to v1.36.1
* moving media transcoder separately
Co-authored-by: Leah Cole <coleleah@google.com>
Co-authored-by: Leah E. Cole <6719667+leahecole@users.noreply.github.com>
* chore(deps): update dependency google-cloud-storage to v1.36.2 (#5520)
* chore(deps): update dependency google-cloud-storage to v1.37.0 (#5580)
[](https://renovatebot.com)
This PR contains the following updates:
| Package | Change | Age | Adoption | Passing | Confidence |
|---|---|---|---|---|---|
| [google-cloud-storage](https://togithub.com/googleapis/python-storage) | `==1.36.2` -> `==1.37.0` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) |
---
### Release Notes
<details>
<summary>googleapis/python-storage</summary>
### [`v1.37.0`](https://togithub.com/googleapis/python-storage/blob/master/CHANGELOG.md#​1370-httpswwwgithubcomgoogleapispython-storagecomparev1362v1370-2021-03-24)
[Compare Source](https://togithub.com/googleapis/python-storage/compare/v1.36.2...v1.37.0)
##### Features
- add blob.open() for file-like I/O ([#​385](https://www.github.com/googleapis/python-storage/issues/385)) ([440a0a4](https://www.github.com/googleapis/python-storage/commit/440a0a4ffe00b1f7c562b0e9c1e47dbadeca33e1)), closes [#​29](https://www.github.com/googleapis/python-storage/issues/29)
##### Bug Fixes
- update user_project usage and documentation in bucket/client class methods ([#​396](https://www.github.com/googleapis/python-storage/issues/396)) ([1a2734b](https://www.github.com/googleapis/python-storage/commit/1a2734ba6d316ce51e4e141571331e86196462b9))
##### [1.36.2](https://www.github.com/googleapis/python-storage/compare/v1.36.1...v1.36.2) (2021-03-09)
##### Bug Fixes
- update batch connection to request api endpoint info from client ([#​392](https://www.github.com/googleapis/python-storage/issues/392)) ([91fc6d9](https://www.github.com/googleapis/python-storage/commit/91fc6d9870a36308b15a827ed6a691e5b4669b62))
##### [1.36.1](https://www.github.com/googleapis/python-storage/compare/v1.36.0...v1.36.1) (2021-02-19)
##### Bug Fixes
- allow metadata keys to be cleared ([#​383](https://www.github.com/googleapis/python-storage/issues/383)) ([79d27da](https://www.github.com/googleapis/python-storage/commit/79d27da9fe842e44a9091076ea0ef52c5ef5ff72)), closes [#​381](https://www.github.com/googleapis/python-storage/issues/381)
- allow signed url version v4 without signed credentials ([#​356](https://www.github.com/googleapis/python-storage/issues/356)) ([3e69bf9](https://www.github.com/googleapis/python-storage/commit/3e69bf92496616c5de28094dd42260b35c3bf982))
- correctly encode bytes for V2 signature ([#​382](https://www.github.com/googleapis/python-storage/issues/382)) ([f44212b](https://www.github.com/googleapis/python-storage/commit/f44212b7b91a67ca661898400fe632f9fb3ec8f6))
</details>
---
### Renovate configuration
:date: **Schedule**: At any time (no schedule defined).
:vertical_traffic_light: **Automerge**: Disabled by config. Please merge this manually once you are satisfied.
:recycle: **Rebasing**: Never, or you tick the rebase/retry checkbox.
:no_bell: **Ignore**: Close this PR and you won't be reminded about this update again.
---
- [ ] <!-- rebase-check -->If you want to rebase/retry this PR, check this box
---
This PR has been generated by [WhiteSource Renovate](https://renovate.whitesourcesoftware.com). View repository job log [here](https://app.renovatebot.com/dashboard#github/GoogleCloudPlatform/python-docs-samples).
* chore(deps): update dependency google-cloud-pubsub to v2.4.1 (#5610)
[](https://renovatebot.com)
This PR contains the following updates:
| Package | Change | Age | Adoption | Passing | Confidence |
|---|---|---|---|---|---|
| [google-cloud-pubsub](https://togithub.com/googleapis/python-pubsub) | `==2.4.0` -> `==2.4.1` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) |
---
### Release Notes
<details>
<summary>googleapis/python-pubsub</summary>
### [`v2.4.1`](https://togithub.com/googleapis/python-pubsub/blob/master/CHANGELOG.md#​241-httpswwwgithubcomgoogleapispython-pubsubcomparev240v241-2021-03-30)
[Compare Source](https://togithub.com/googleapis/python-pubsub/compare/v2.4.0...2.4.1)
##### Bug Fixes
- Move `await_msg_callbacks` flag to `subscribe()` method, fixing a regression in Pub/Sub Lite client.
([#​320](https://www.github.com/googleapis/python-pubsub/issues/320)) ([d40d027](https://www.github.com/googleapis/python-pubsub/commit/d40d02713c8c189937ae5c21d099b88a3131a59f))
- SSL error when using the client with the emulator. ([#​297](https://www.github.com/googleapis/python-pubsub/issues/297)) ([83db672](https://www.github.com/googleapis/python-pubsub/commit/83db67239d3521457138699109f766d574a0a2c4))
##### Implementation Changes
- (samples) Bump the max_time to 10 minutes for a flaky test. ([#​311](https://www.github.com/googleapis/python-pubsub/issues/311)) ([e2678d4](https://www.github.com/googleapis/python-pubsub/commit/e2678d47c08e6b03782d2d744a4e630b933fdd51)), closes [#​291](https://www.github.com/googleapis/python-pubsub/issues/291)
- (samples) Mark delivery attempts test as flaky. ([#​326](https://www.github.com/googleapis/python-pubsub/issues/326)) ([5a97ef1](https://www.github.com/googleapis/python-pubsub/commit/5a97ef1bb7512fe814a8f72a43b3e9698434cd8d))
- (samples) Mitigate flakiness in subscriber_tests. ([#​304](https://www.github.com/googleapis/python-pubsub/issues/304)) ([271a385](https://www.github.com/googleapis/python-pubsub/commit/271a3856d835967f18f6becdae5ad53d585d0ccf))
- (samples) Retry `InternalServerError` in dead letter policy test. ([#​329](https://www.github.com/googleapis/python-pubsub/issues/329)) ([34c9b11](https://www.github.com/googleapis/python-pubsub/commit/34c9b11ae697c280f32642c3101b7f7da971f589)), closes [#​321](https://www.github.com/googleapis/python-pubsub/issues/321)
##### Documentation
- Remove EXPERIMENTAL tag for ordering keys in `types.py`. ([#​323](https://www.github.com/googleapis/python-pubsub/issues/323)) ([659cd7a](https://www.github.com/googleapis/python-pubsub/commit/659cd7ae2784245d4217fbc722dac04bd3222d32))
- Remove EXPERIMENTAL tag from `Schema` service (via synth). ([#​307](https://www.github.com/googleapis/python-pubsub/issues/307)) ([ad85202](https://www.github.com/googleapis/python-pubsub/commit/ad852028836520db779c5cc33689ffd7e5458a7d))
</details>
---
### Renovate configuration
:date: **Schedule**: At any time (no schedule defined).
:vertical_traffic_light: **Automerge**: Disabled by config. Please merge this manually once you are satisfied.
:recycle: **Rebasing**: Never, or you tick the rebase/retry checkbox.
:no_bell: **Ignore**: Close this PR and you won't be reminded about this update again.
---
- [ ] <!-- rebase-check -->If you want to rebase/retry this PR, check this box
---
This PR has been generated by [WhiteSource Renovate](https://renovate.whitesourcesoftware.com). View repository job log [here](https://app.renovatebot.com/dashboard#github/GoogleCloudPlatform/python-docs-samples).
* docs: update description of parameters in storage_upload_file (#5707)
* following Java's example https://github.com/googleapis/google-cloud-java/blob/b36db6a957bcfb7b6ccdb77fb12b4cc7fa22b807/google-cloud-examples/src/main/java/com/google/cloud/examples/storage/objects/UploadObject.java#L33-L40
* samples(storage): update storage_set_bucket_public_iam to explicitly set role and member (#5708)
* chore: fix typo on noxfile (#5739)
* chore: add noxfile config
* chore: fix typo on noxfile
* Remove "chore: add noxfile config"
This reverts commit 61972125cbbf110941da1227afed53f169bad3a6.
* chore: fix the base noxfile_config
* fix(storage): retry flaky test (#5744)
Fixes #5684
* chore(deps): update dependency google-api-python-client to v2.3.0 (#5689)
* Update storage_list_files_with_prefix.py (#5747)
* chore(deps): update dependency google-cloud-storage to v1.38.0 (#5640)
Test failures are unrelated
* chore(deps): update dependency pytest to v6.2.4 (#5787)
Co-authored-by: Dan Lee <71398022+dandhlee@users.noreply.github.com>
* chore(deps): update dependency google-cloud-pubsub to v2.4.2 (#5810)
Co-authored-by: Dan Lee <71398022+dandhlee@users.noreply.github.com>
* chore(deps): update dependency google-cloud-pubsub to v2.5.0 (#5845)
* chore(deps): update dependency google-api-python-client to v2.4.0 (#5820)
* chore(deps): update dependency google-api-python-client to v2.5.0 (#5857)
Co-authored-by: Dan Lee <71398022+dandhlee@users.noreply.github.com>
* chore(deps): update dependency google-api-python-client to v2.6.0 (#5890)
Co-authored-by: Dan Lee <71398022+dandhlee@users.noreply.github.com>
* chore(deps): update dependency google-api-python-client to v2.7.0 (#6062)
* chore(deps): update dependency google-cloud-pubsub to v2.6.0 (#6233)
* public access prevention samples & tests (#4971)
* public access prevention samples & tests
* linted files
* respnded to PR comments
* updated docstring
* updated docstring
* refactored fixture code
* renamed samples
* updated location for constants
* updated location for constants
* updated samples to conform to sample guidelines
* added license
* updated headers
* Updating requirements
* used f strings
* linted files
* f string suggestions from code review
Co-authored-by: Dina Graves Portman <dinagraves@google.com>
Co-authored-by: Dina Graves Portman <dinagraves@google.com>
* chore(deps): update dependency google-api-python-client to v2.11.0 (#6101)
* chore(deps): update dependency google-cloud-pubsub to v2.6.1 (#6284)
* chore(deps): update dependency backoff to v1.11.0 (#6285)
Co-authored-by: Dina Graves Portman <dinagraves@google.com>
* chore(deps): update dependency google-cloud-storage to v1.41.0 (#6197)
* chore(deps): update dependency google-cloud-storage to v1.41.0
* revert dataflow flex templates
* revert all dataflow changes
* correct my mistake with dataflow stuff
* restore dataflow file
Co-authored-by: Leah Cole <coleleah@google.com>
* chore(deps): update dependency google-api-python-client to v2.12.0 (#6269)
* chore(deps): update dependency google-api-python-client to v2.12.0
* revert dataflow
* revert dataflow
Co-authored-by: Leah Cole <coleleah@google.com>
* chore(deps): update dependency google-cloud-pubsub to v2.7.0 (#6486)
Co-authored-by: Dan Lee <71398022+dandhlee@users.noreply.github.com>
Co-authored-by: Anthonios Partheniou <partheniou@google.com>
* fix(storage): update service account email for acl tests (#6529)
* fix: update test email for acl tests. previous email was deleted in the project
* update to service account without project editor owner permissions
* update test email to avoid creating new service accounts
* docs(storage): update description in storage_download_file (#6553)
* Add storage move_blob sample and fix confusion with rename (#6554)
* Add storage move_blob sample and fix confusion with rename
* fix license heading issues
* Add descriptive comments to parameters
* Update storage/cloud-client/storage_move_file.py
* Apply suggestions from code review
Add print statement in except clause
Co-authored-by: cojenco <cathyo@google.com>
Co-authored-by: Leah E. Cole <6719667+leahecole@users.noreply.github.com>
* chore(deps): update dependency backoff to v1.11.1 (#6571)
Co-authored-by: Leah E. Cole <6719667+leahecole@users.noreply.github.com>
* chore(deps): update dependency google-api-python-client to v2.15.0 (#6574)
[](https://renovatebot.com)
This PR contains the following updates:
| Package | Change | Age | Adoption | Passing | Confidence |
|---|---|---|---|---|---|
| [google-api-python-client](https://togithub.com/googleapis/google-api-python-client) | `==2.12.0` -> `==2.15.0` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) |
| [google-api-python-client](https://togithub.com/googleapis/google-api-python-client) | `==2.11.0` -> `==2.15.0` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) |
| [google-api-python-client](https://togithub.com/googleapis/google-api-python-client) | `==2.1.0` -> `==2.15.0` | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) | [](https://docs.renovatebot.com/merge-confidence/) |
---
### Release Notes
<details>
<summary>googleapis/google-api-python-client</summary>
### [`v2.15.0`](https://togithub.com/googleapis/google-api-python-client/blob/master/CHANGELOG.md#​2150-httpswwwgithubcomgoogleapisgoogle-api-python-clientcomparev2141v2150-2021-07-27)
[Compare Source](https://togithub.com/googleapis/google-api-python-client/compare/v2.14.1...v2.15.0)
##### Features
- **alertcenter:** update the api https://github.com/googleapis/google-api-python-client/commit/70810a52c85c6d0d6f00d7afb41c8608261eaebc ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **chat:** update the api https://github.com/googleapis/google-api-python-client/commit/a577cd0b71951176bbf849c1f7f139127205da54 ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **cloudbuild:** update the api https://github.com/googleapis/google-api-python-client/commit/9066056a8b106d441fb7686fe84359484d0d58bc ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **content:** update the api https://github.com/googleapis/google-api-python-client/commit/b123349da33c11c0172a8efb3fadef685a30e6e1 ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **displayvideo:** update the api https://github.com/googleapis/google-api-python-client/commit/c525d726ee6cffdd4bc7afd69080d5e52bae83a0 ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **dns:** update the api https://github.com/googleapis/google-api-python-client/commit/13436ccd2b835fda5cb86952ac4ea991ee8651d8 ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **eventarc:** update the api https://github.com/googleapis/google-api-python-client/commit/6be3394a64a5eb509f68ef779680fd36837708ee ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **file:** update the api https://github.com/googleapis/google-api-python-client/commit/817a0e636771445a988ef479bd52740f754b901a ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **monitoring:** update the api https://github.com/googleapis/google-api-python-client/commit/bd32149f308467f0f659119587afc77dcec65b14 ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **people:** update the api https://github.com/googleapis/google-api-python-client/commit/aa6b47df40c5289f33aef6fb6aa007df2d038e20 ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **retail:** update the api https://github.com/googleapis/google-api-python-client/commit/d39f06e2d77034bc837604a41dd52c577f158bf2 ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **securitycenter:** update the api https://github.com/googleapis/google-api-python-client/commit/999fab5178208639c9eef289f9f441052ed832fc ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **speech:** update the api https://github.com/googleapis/google-api-python-client/commit/3b2c0fa62b2a0c86bba1e97f1b18f93250dbd551 ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfdc5b3ac0a70ff1b014c4bc75f))
- **sqladmin:** update the api https://github.com/googleapis/google-api-python-client/commit/cef24d829ab5be71563a2b668b8f6cf5dda2c8e4 ([a36e3b1](https://www.github.com/googleapis/google-api-python-client/commit/a36e3b130d609dfd…
- **Add methods for Models API (#7562)** - **Move table and dataset snippets to samples/ directory (#7683)** - **Add sample demonstrating how to create a job. (#8422)** - **Add Routines API. (#8491)** - **Add `to_arrow` to get a `pyarrow.Table` from query results. (#8609)** - **BigQuery: deprecate `client.dataset()` part 1 (#9032)** - **Optionally include indexes in table written by `load_table_from_dataframe`. (#9084)** - **docs(bigquery): standardize comments and formatting in existing code samples (#9212)** - **feat(bigquery): add support for sheets ranges (#9416)** - **refactor(bigquery): update code samples to use strings for table and dataset IDs (#9136)** - **feat(bigquery): add Dataset.default_partition_expiration_ms and Table.require_partition_filter properties (#9464)** - **doc(bigquery): add code sample for scripting (#9537)** - **chore(bigquery): format code with latest version of black (#9556)** - **docs(bigquery): document how to achieve higher write limit and add tests (#9574)** - **doc(bigquery): add table create sample using integer range partitioning (#9478)** - **chore(bigquery): fix undelete table system test to use milliseconds in snapshot decorator (#9649)** - **refactor(bigquery): update code samples to use strings for table and dataset IDs (#9495)** - **docs(bigquery): add sample to read from query destination table (#9964)** - **feat(bigquery): add `create_bqstorage_client` param to `to_dataframe` and `to_arrow` (#9573)** - **refactor(bigquery): update code samples to use strings for table and dataset IDs (#9974)** - **cleanup(bigquery): remove unused query code sample (#10024)** - **doc(bigquery): uncomment `Client` constructor and imports in samples (#10058)** - **fix(bigquery): write pandas `datetime[ns]` columns to BigQuery TIMESTAMP columns (#10028)** - **refactor(bigquery): update code samples of load table file and uri (#10175)** - **docs: use timeout in create dataset sample (#52)** - **refactor(bigquery): update code samples of load table autodetect and truncate (#28)** - **Add veneer samples for BigQuery. [(#444)](GoogleCloudPlatform/python-docs-samples#444 - **Add data loading samples** - **Add bigquery snippets** - **Remove unnecessary fixture** - **Address review comments** - **Fix lint** - **Address post-review comments for BigQuery samples** - **Add bigquery create and copy table examples [(#514)](GoogleCloudPlatform/python-docs-samples#514 - **Auto-update dependencies. [(#540)](GoogleCloudPlatform/python-docs-samples#540 - **Auto-update dependencies. [(#542)](GoogleCloudPlatform/python-docs-samples#542 - **Move to google-cloud [(#544)](GoogleCloudPlatform/python-docs-samples#544 - **Add new "quickstart" samples [(#547)](GoogleCloudPlatform/python-docs-samples#547 - **BigQuery - use new QueryJob.results() method. [(#555)](GoogleCloudPlatform/python-docs-samples#555 - **Quickstart tests [(#569)](GoogleCloudPlatform/python-docs-samples#569 - **Generate readmes for most service samples [(#599)](GoogleCloudPlatform/python-docs-samples#599 - **Generate most non-appengine readmes** - **Add missing bigquery samples [(#622)](GoogleCloudPlatform/python-docs-samples#622 - **Update samples to support latest Google Cloud Python [(#656)](GoogleCloudPlatform/python-docs-samples#656 - **Update readmes** - **Fix bigquery load from file sample** - **Fix lint** - **Adjusted error handling based on Googler feedback [(#693)](GoogleCloudPlatform/python-docs-samples#693 - **Changed error handling for all other big query samples as well [(#694)](GoogleCloudPlatform/python-docs-samples#694 - **Add bigquery shakespeare example [(#604)](GoogleCloudPlatform/python-docs-samples#604 - **BigQuery parameterized query sample [(#699)](GoogleCloudPlatform/python-docs-samples#699 - **Auto-update dependencies. [(#715)](GoogleCloudPlatform/python-docs-samples#715 - **Remove cloud config fixture [(#887)](GoogleCloudPlatform/python-docs-samples#887 - **Remove resource [(#890)](GoogleCloudPlatform/python-docs-samples#890 - **BigQuery: named parameter query samples** - **Rename query file and use async method.** - **Indent queries.** - **Auto-update dependencies. [(#916)](GoogleCloudPlatform/python-docs-samples#916 - **Re-generate all readmes** - **BigQuery: user credentials to run a query. [(#925)](GoogleCloudPlatform/python-docs-samples#925 - **Auto-update dependencies. [(#942)](GoogleCloudPlatform/python-docs-samples#942 - **Fix README rst links [(#962)](GoogleCloudPlatform/python-docs-samples#962 - **Auto-update dependencies. [(#1004)](GoogleCloudPlatform/python-docs-samples#1004 - **BigQuery: add auth samples for service accounts.** - **Fix SQL query to return 10 rows & simplify [(#1041)](GoogleCloudPlatform/python-docs-samples#1041 - **Fix test randomness [(#1043)](GoogleCloudPlatform/python-docs-samples#1043 - **Fix test randomness [(#1044)](GoogleCloudPlatform/python-docs-samples#1044 - **Auto-update dependencies. [(#1055)](GoogleCloudPlatform/python-docs-samples#1055 - **Update readme references. Resolves 1059** - **Use futures API to wait for jobs to finish.** - **BQ: Use futures API for quickstart.** - **BigQuery : max_results changes page size not full list size** - **Remove max_results argument from table.fetch_data() call.** - **Remove unnecessary list() call** - **Auto-update dependencies. [(#1093)](GoogleCloudPlatform/python-docs-samples#1093 - **BigQuery : add sample for writing query results to a destination table. [(#1101)](GoogleCloudPlatform/python-docs-samples#1101 - **Auto-update dependencies. [(#1128)](GoogleCloudPlatform/python-docs-samples#1128 - **BigQuery: Updates samples for BigQuery Beta 2 (do not merge until release) [(#1178)](GoogleCloudPlatform/python-docs-samples#1178 - **Auto-update dependencies. [(#1186)](GoogleCloudPlatform/python-docs-samples#1186 - **Auto-update dependencies. [(#1227)](GoogleCloudPlatform/python-docs-samples#1227 - **/s/buckets/datasets in BigQuery auth sample [(#1242)](GoogleCloudPlatform/python-docs-samples#1242 - **Fix broken link to core Client service account helper. [(#1256)](GoogleCloudPlatform/python-docs-samples#1256 - **Added "Open in Cloud Shell" buttons to README files [(#1254)](GoogleCloudPlatform/python-docs-samples#1254 - **Move imports into bigquery samples** - **BigQuery: rewrite simple app tutorial.** - **Auto-update dependencies. [(#1272)](GoogleCloudPlatform/python-docs-samples#1272 - **Auto-update dependencies. [(#1307)](GoogleCloudPlatform/python-docs-samples#1307 - **Auto-update dependencies. [(#1355)](GoogleCloudPlatform/python-docs-samples#1355 - **Auto-update dependencies. [(#1377)](GoogleCloudPlatform/python-docs-samples#1377 - **Regenerate the README files and fix the Open in Cloud Shell link for some samples [(#1441)](GoogleCloudPlatform/python-docs-samples#1441 - **BigQuery: Remove unused samples [(#1455)](GoogleCloudPlatform/python-docs-samples#1455 - **BigQuery: Removes unused API samples [(#1459)](GoogleCloudPlatform/python-docs-samples#1459 - **Update READMEs to fix numbering and add git clone [(#1464)](GoogleCloudPlatform/python-docs-samples#1464 - **BigQuery: Moves BigQuery tutorial for Dataproc to python-docs-samples [(#1494)](GoogleCloudPlatform/python-docs-samples#1494 - **BigQuery: Adds authorized view tutorial [(#1535)](GoogleCloudPlatform/python-docs-samples#1535 - **Region tag end typo [(#1536)](GoogleCloudPlatform/python-docs-samples#1536 - **Code snippets for Jupyter data visualization tutorial [(#1560)](GoogleCloudPlatform/python-docs-samples#1560 - **updates last query per Steve's feedback [(#1578)](GoogleCloudPlatform/python-docs-samples#1578 - **updates tutorial to add datasets to to_delete before running [(#1579)](GoogleCloudPlatform/python-docs-samples#1579 - **Remove ipython Python 2 modifier from requirements.txt [(#1675)](GoogleCloudPlatform/python-docs-samples#1675 - **Auto-update dependencies. [(#1846)](GoogleCloudPlatform/python-docs-samples#1846 - **Refactor BQ user credentials sample to use region tags [(#1952)](GoogleCloudPlatform/python-docs-samples#1952 - **Make TODO clearer what action to take. [(#1963)](GoogleCloudPlatform/python-docs-samples#1963 - **Add missing explanation about local_server callback URL** - **Auto-update dependencies. [(#1980)](GoogleCloudPlatform/python-docs-samples#1980 - **BigQuery: Add sample for explicitly creating client from service acco… [(#2312)](GoogleCloudPlatform/python-docs-samples#2312 - **Adds updates including compute [(#2436)](GoogleCloudPlatform/python-docs-samples#2436 - **Simplify noxfile setup. [(#2806)](GoogleCloudPlatform/python-docs-samples#2806 - **fix: address issues with concurrent BigQuery tests [(#3426)](GoogleCloudPlatform/python-docs-samples#3426 - **chore(deps): update dependency pytz to v2019.3 [(#3190)](GoogleCloudPlatform/python-docs-samples#3190 - **chore(deps): update dependency matplotlib to v3.2.1 [(#3180)](GoogleCloudPlatform/python-docs-samples#3180 - **chore(deps): update dependency ipython to v7.13.0 [(#3177)](GoogleCloudPlatform/python-docs-samples#3177 - **chore(deps): update dependency google-cloud-bigquery to v1.24.0 [(#3049)](GoogleCloudPlatform/python-docs-samples#3049 - **chore(deps): update dependency pytz to v2020 [(#3566)](GoogleCloudPlatform/python-docs-samples#3566 - **chore(deps): update dependency ipython to v7.14.0 [(#3670)](GoogleCloudPlatform/python-docs-samples#3670 - **[bigquery] opt in to use the build specific projects [(#3883)](GoogleCloudPlatform/python-docs-samples#3883 - **chore(deps): update dependency ipython to v7.15.0 [(#3917)](GoogleCloudPlatform/python-docs-samples#3917 - **Replace GCLOUD_PROJECT with GOOGLE_CLOUD_PROJECT. [(#4022)](GoogleCloudPlatform/python-docs-samples#4022 - **Update dependency google-cloud-bigquery to v1.25.0 [(#4024)](GoogleCloudPlatform/python-docs-samples#4024 - **chore(deps): update dependency matplotlib to v3.2.2 [(#4118)](GoogleCloudPlatform/python-docs-samples#4118 - **Update dependency ipython to v7.16.1 [(#4189)](GoogleCloudPlatform/python-docs-samples#4189 - **chore(deps): update dependency pytest to v5.4.3 [(#4279)](GoogleCloudPlatform/python-docs-samples#4279 - **chore(deps): update dependency mock to v4 [(#4287)](GoogleCloudPlatform/python-docs-samples#4287 - **chore(deps): update dependency matplotlib to v3.3.0 [(#4310)](GoogleCloudPlatform/python-docs-samples#4310 - **chore: update templates** - **chore: change project env variable to use in tests** - **chore: fix samples tests dependencies** - **chore(deps): update dependency google-cloud-bigquery to v1.26.1 (#180)** - **chore(deps): pin ipython to v7.17.0 for samples, python >= 3.7 (#208)** - **chore(deps): update dependency matplotlib to v3.3.1 (#224)** - **docs: recommend insert_rows_json to avoid call to tables.get (#258)** - **fix: update minimum dependency versions (#263)** - **chore: Prepare for 2.0 release (#278)** - **chore(deps): update dependency google-cloud-bigquery to v2 (#287)** - **docs(samples): add create_table_clustered code snippet (#291)** - **docs: update snippets samples to support version 2.0 (#309)** - **chore(deps): update dependency google-cloud-bigquery to v2.1.0 (#312)** - **chore(deps): update dependency matplotlib to v3.3.2 (#260)** - **chore(deps): update dependency grpcio to v1.33.1 (#338)** - **chore(deps): update dependency google-auth-oauthlib to v0.4.2 (#349)** - **chore(deps): update dependency grpcio to v1.33.2 (#350)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.0.1 (#337)** - **chore(deps): update dependency google-cloud-bigquery to v2.2.0 (#333)** - **chore(deps): update dependency pandas to v1.1.4 (#355)** - **docs(samples): add more clustering code snippets (#330)** - **chore(deps): update dependency matplotlib to v3.3.3 (#385)** - **chore(deps): update dependency google-cloud-bigquery to v2.3.1 (#371)** - **chore(deps): update dependency google-cloud-bigquery to v2.4.0 (#391)** - **chore(deps): update dependency pyarrow to v2 (#380)** - **chore(deps): update dependency pytz to v2020.4 (#356)** - **feat: add support for materialized views (#408)** - **chore(deps): update dependency grpcio to v1.34.0 (#411)** - **chore(deps): update dependency google-cloud-bigquery to v2.5.0 (#410)** - **docs: update intersphinx links (#404)** - **chore(deps): update dependency google-cloud-bigquery to v2.6.0 (#419)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.1.0 (#369)** - **chore(deps): update dependency pandas to v1.1.5 (#417)** - **docs: add GEOGRAPHY data type code samples (#428)** - **docs: fix Shapely import in GEOGRAPHY sample (#431)** - **chore(deps): update dependency google-cloud-bigquery to v2.6.1 (#430)** - **docs: move and refresh view samples (#420)** - **chore(deps): update dependency pytz to v2020.5 (#452)** - **chore(deps): update dependency pandas to v1.2.0 (#454)** - **chore: remove six dependency (#461)** - **chore(deps): update dependency google-cloud-bigquery to v2.6.2 (#466)** - **chore(deps): update dependency grpcio to v1.34.1 (#470)** - **chore(deps): update dependency grpcio to v1.35.0 (#477)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.2.1 (#485)** - **chore(deps): update dependency pyarrow to v3 (#490)** - **chore(deps): update dependency google-cloud-bigquery to v2.7.0 (#491)** - **chore(deps): update dependency matplotlib to v3.3.4 (#495)** - **chore: update shared templates (#468)** - **chore(deps): update dependency pytz to v2021 (#502)** - **test: add samples Python 3.9 test session (#506)** - **docs(bigquery): Add alternative approach to setting credentials (#517)** - **chore(deps): update dependency google-cloud-bigquery to v2.9.0 (#515)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.3.0 (#529)** - **chore(deps): update dependency google-cloud-bigquery to v2.10.0 (#535)** - **chore(deps): update dependency grpcio to v1.36.0 (#536)** - **chore(deps): update dependency grpcio to v1.36.1 (#541)** - **chore(deps): update dependency google-auth-oauthlib to v0.4.3 (#542)** - **chore(deps): update dependency google-cloud-bigquery to v2.11.0 (#546)** - **chore(deps): update dependency google-cloud-bigquery to v2.12.0 (#554)** - **chore(deps): update dependency google-cloud-bigquery to v2.13.0 (#570)** - **chore(deps): update dependency google-auth-oauthlib to v0.4.4 (#578)** - **chore(deps): update dependency matplotlib to v3.4.1 (#576)** - **chore(deps): update dependency grpcio to v1.37.0 (#596)** - **chore: add constraints file check for python samples (#601)** - **chore(deps): update dependency mock to v4.0.3 (#605)** - **docs: add sample to run DML query (#591)** - **chore(deps): update dependency pytest to v6 (#606)** - **chore(deps): update dependency google-cloud-bigquery to v2.13.1 (#573)** - **feat: add type hints for public methods (#613)** - **chore(deps): update dependency google-cloud-bigquery to v2.14.0 (#627)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.4.0 (#595)** - **chore(deps): update dependency google-cloud-bigquery to v2.15.0 (#639)** - **chore(deps): update dependency pyarrow to v4 (#641)** - **chore(deps): update dependency grpcio to v1.37.1 (#640)** - **chore(deps): update dependency pytest to v6.2.4 (#647)** - **chore(deps): update dependency google-cloud-bigquery to v2.16.1 (#652)** - **chore(deps): update dependency grpcio to v1.38.0 (#674)** - **fix(tests): invalid path to strptime() (#672)** - **chore(deps): update dependency google-cloud-bigquery to v2.17.0 (#675)** - **chore: new owl bot post processor docker image (#677)** - **chore(deps): update dependency pyarrow to v4.0.1 (#681)** - **chore(deps): update dependency google-cloud-bigquery to v2.18.0 (#683)** - **chore(deps): update dependency google-cloud-bigquery to v2.19.0 (#691)** - **chore(deps): update dependency google-cloud-bigquery to v2.20.0 (#694)** - **chore(deps): update dependency grpcio to v1.38.1 (#713)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.5.0 (#731)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.6.0 (#743)** - **chore(deps): update dependency google-cloud-bigquery to v2.21.0 (#755)** - **docs: add loading data from Firestore backup sample (#737)** - **chore(deps): update dependency google-cloud-bigquery to v2.22.0 (#783)** - **feat: add Samples section to CONTRIBUTING.rst (#785)** - **docs: add sample to delete job metadata (#798)** - **chore(deps): update dependency grpcio to v1.39.0 (#796)** - **chore(deps): update dependency google-cloud-bigquery to v2.22.1 (#800)** - **chore(deps): update dependency google-cloud-bigquery to v2.23.0 (#820)** - **chore(deps): update dependency google-cloud-bigquery to v2.23.1 (#827)** - **chore(deps): update dependency pyarrow to v5 (#834)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.6.2 (#795)** - **chore(deps): update dependency google-auth-oauthlib to v0.4.5 (#839)** - **chore(deps): update dependency google-cloud-bigquery to v2.23.2 (#838)** - **chore(deps): update dependency google-cloud-testutils to v1 (#845)** - **chore(deps): update dependency google-cloud-bigquery to v2.23.3 (#866)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.6.3 (#863)** - **chore: fix INSTALL_LIBRARY_FROM_SOURCE in noxfile.py (#869)** - **chore(deps): update dependency google-cloud-bigquery to v2.24.0 (#873)** - **chore: drop mention of Python 2.7 from templates (#877)** - **fix: remove pytz dependency and require pyarrow>=3.0.0 (#875)** - **chore(deps): update dependency google-cloud-bigquery to v2.24.1 (#887)** - **feat: Support using GeoPandas for GEOGRAPHY columns (#848)** - **chore(deps): update dependency numpy to v1.21.2 (#899)** - **chore(deps): update dependency google-cloud-core to v2 (#904)** - **chore(deps): update dependency google-cloud-bigquery to v2.25.0 (#907)** - **chore(deps): update dependency pandas to v1.3.2 (#900)** - **docs: update docstring for bigquery_create_routine sample (#883) (#917)** - **chore(deps): update all dependencies (#914)** - **chore(deps): update all dependencies (#926)** - **chore(deps): update all dependencies (#928)** - **chore(deps): update dependency google-cloud-bigquery to v2.26.0 (#938)** - **chore: update system tests and samples to use and @google.com email address (#942)** - **test: fix routine DDL sample test exits too early (#932)** - **chore(deps): update all dependencies (#939)** - **chore(deps): update all dependencies (#943)** - **docs: simplify destination table sample with f-strings (#966)** - **chore(deps): update all dependencies (#963)** - **chore(deps): update all dependencies (#992)** - **chore: fail samples nox session if python version is missing (#997)** - **chore(deps): update all dependencies (#993)** - **chore(python): Add kokoro configs for python 3.10 samples testing (#1010)** - **docs(samples): add create external table with hive partitioning (#1033)** - **docs: add code samples for Jupyter/IPython magics (#1013)** - **chore: use gapic-generator-python 0.53.4 (#1037)** - **feat: support Python 3.10 (#1043)** - **docs: show gcloud command to authorize against sheets (#1045)** - **docs: add sample for revoking dataset access (#778)** - **chore: update python-docs-samples link to main branch (#1082)** - **feat: support authorized dataset entity (#1075)** - **chore(deps): update all dependencies (#1005)** - **test: skip failing copy table tests (#1090)** - **chore(deps): update all dependencies (#1089)** - **chore(deps): update all dependencies (#1091)** - **chore(deps): update all dependencies (#1098)** - **chore(samples): Add check for tests in directory (#1103)** - **chore(deps): update all dependencies (#1101)** - **chore(python): Noxfile recognizes that tests can live in a folder (#1115)** - **docs: show common job properties in `get_job` and `cancel_job` samples (#1137)** - **chore(deps): update dependency ipython to v8.0.1 [security] (#1126)** - **chore(deps): update dependency ipython [security] (#1154)** - **samples: remove browser option from user credentials sample (#1147)** - **deps: allow pyarrow 7.0 (#1112)** - **chore: Adding support for pytest-xdist and pytest-parallel (#1160)** - **chore(deps): update all dependencies (#1156)** - **chore(deps): update all dependencies (#1162)** - **chore(deps): update all dependencies (#1163)** - **fix: upgrade black (#1177)** - **fix!: remove out-of-date BigQuery ML protocol buffers (#1178)** - **chore(deps): update all dependencies (#1164)** - **chore(deps): update all dependencies (#1183)** - **chore: use === in requirements.txt for environment specific pins (#1184)** - **chore(deps): update all dependencies (#1185)** - **chore(deps): update all dependencies (#1192)** - **chore(deps): update dependency click to v8.1.2 (#1197)** - **chore(deps): update all dependencies (#1200)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.13.1 (#1203)** - **chore(deps): update dependency google-auth to v2.6.3 (#1206)** - **chore(deps): update dependency pyparsing to v3.0.8 (#1210)** - **chore(deps): update dependency google-cloud-core to v2.3.0 (#1211)** - **chore(deps): update dependency google-auth to v2.6.4 (#1214)** - **chore(deps): update dependency google-api-core to v2.7.2 (#1215)** - **fix: export bigquery.HivePartitioningOptions (#1217)** - **chore(deps): update dependency google-auth to v2.6.5 (#1219)** - **chore(deps): update dependency typing-extensions to v4.2.0 (#1221)** - **chore(python): add nox session to sort python imports (#1224)** - **chore(deps): update all dependencies (#1227)** - **chore(deps): update dependency pytest to v7.1.2 (#1229)** - **chore(deps): update dependency click to v8.1.3 (#1233)** - **chore(deps): update dependency ipython to v8.3.0 (#1234)** - **chore(deps): update dependency google-api-core to v2.7.3 (#1236)** - **chore(deps): update dependency matplotlib to v3.5.2 (#1237)** - **chore(deps): update all dependencies (#1238)** - **docs: updated variable typo in comment in code sample (#1239)** - **chore: update all dependencies (#1244)** - **chore(deps): update dependency pyparsing to v3.0.9 (#1246)** - **chore(deps): update dependency libcst to v0.4.3 (#1247)** - **chore(deps): update dependency grpcio to v1.46.1 (#1249)** - **chore(deps): update dependency googleapis-common-protos to v1.56.1 (#1251)** - **chore(deps): update all dependencies (#1253)** - **chore(deps): update all dependencies (#1255)** - **feat: add support for table clones (#1235)** - **chore(deps): update all dependencies (#1258)** - **chore(deps): update all dependencies (#1264)** - **docs(samples): add table snapshot sample (#1274)** - **fix: require python 3.7+ (#1284)** - **docs(samples): explicitly add bq to samples reqs, upgrade grpc to fix bug on m1 (#1290)** - **chore(deps): update all dependencies (#1301)** - **chore(deps): update all dependencies (#1305)** - **chore(deps): update all dependencies (#1308)** - **chore(deps): update all dependencies (#1313)** - **chore(deps): update dependency pytz to v2022.2.1 (#1316)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.14.2 (#1317)** - **chore(deps): update dependency shapely to v1.8.3 (#1320)** - **chore(deps): update all dependencies (#1322)** - **chore(deps): update all dependencies (#1323)** - **chore(deps): update dependency proto-plus to v1.22.1 (#1330)** - **chore(deps): update dependency pandas to v1.4.4 (#1332)** - **chore(deps): update dependency google-crc32c to v1.5.0 (#1334)** - **fix: uses function (create_job) more appropriate to the described sample intent (#1309)** - **chore(deps): update all dependencies (#1339)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.15.0 (#1342)** - **chore(deps): update dependency certifi to v2022.6.15.1 (#1346)** - **chore: detect samples tests in nested directories (#1349)** - **chore(deps): update all dependencies (#1351)** - **chore(deps): update dependency google-api-core to v2.10.1 (#1353)** - **chore(deps): update all dependencies (#1354)** - **chore(deps): update all dependencies (#1358)** - **chore(deps): update all dependencies (#1360)** - **chore(deps): update dependency db-dtypes to v1.0.4 (#1361)** - **chore(deps): update all dependencies (#1363)** - **chore(deps): update all dependencies (#1380)** - **feat: reconfigure tqdm progress bar in %%bigquery magic (#1355)** - **chore(deps): update all dependencies (#1382)** - **fix: corrects test for non-existent attribute (#1395)** - **chore(python): drop flake8-import-order in samples noxfile (#1424)** - **docs: created samples for load table and create table from schema file (#1436)** - **docs: create sample to write schema file from table (#1439)** - **docs: revise sample for nested schema (#1446)** - **chore: install all sample dependencies when type checking samples (#1455)** - **docs: revise label table code samples (#1451)** - **docs: revise create table cmek sample (#1452)** - **chore(python): add support for python 3.11 [autoapprove] (#1454)** - **chore(deps): update dependency certifi to v2022.12.7 [security] (#1434)** - **chore(deps): update all dependencies (#1390)** - **chore(deps): update all dependencies (#1456)** - **chore(deps): update all dependencies (#1460)** - **chore(deps): update all dependencies (#1461)** - **docs: revise get table labels code sample, add TODO to clean up snipp… (#1464)** - **chore(deps): update dependency packaging to v23 (#1462)** - **docs: revise delete label table code sample, add TODO to clean up sni… (#1466)** - **docs: adds snippet for creating table with external data config (#1420)** - **chore(deps): update all dependencies (#1471)** - **chore(deps): update all dependencies (#1475)** - **docs(samples): table variable fix (#1287)** - **chore(deps): update dependency google-cloud-bigquery-storage to v2.18.1 (#1476)** - **chore(deps): update dependency geojson to v3 (#1478)** - **chore(deps): update all dependencies (#1479)** - **chore(deps): update dependency shapely to v2.0.1 (#1480)** - **chore(deps): update dependency fiona to v1.9.0 (#1482)** - **chore(deps): update dependency google-cloud-bigquery to v3.5.0 (#1485)** - **chore(deps): update all dependencies to v1 (#1486)** - **chore(deps): update all dependencies (#1501)** - **chore(deps): update all dependencies (#1513)** - **chore(deps): update all dependencies (#1514)** - **chore(deps): update dependency charset-normalizer to v3.1.0 (#1518)** - **chore(deps): update all dependencies (#1522)** - **chore(deps): update all dependencies (#1540)** - **chore(deps): update all dependencies (#1549)** - **chore(deps): update dependency google-cloud-bigquery to v3.10.0 (#1556)** - **chore(deps): update all dependencies (#1557)** - **chore(deps): update dependency requests to v2.31.0 [security] (#1573)** - **chore(deps): update all dependencies (#1560)** - **chore(deps): update all dependencies (#1576)** - **chore(deps): update all dependencies (#1577)** - **chore(deps): update all dependencies (#1606)** - **test: enable copy table tests (#1609)** - **chore(deps): update all dependencies (#1607)** - **chore(deps): bump certifi in /samples/geography (#1627)** - **chore(deps): update all dependencies (#1626)** - **chore(deps): update all dependencies (#1636)** - **chore(docs): update region tag and move sample and test (#1648)** - **docs: Revise update_table_expiration sample (#1457)** - **docs: revised `create_partitioned_table` sample (#1447)** - **docs: revised relax column mode sample (#1467)** - **chore: [autoapprove] Update `black` and `isort` to latest versions (#1678)** - **chore: upgrade mypy (#1705)** - **chore(deps): bump urllib3 from 1.26.15 to 1.26.18 in /samples/geography (#1684)** - **chore(deps): update dependency pyarrow to v14 [security] (#1718)** - **chore: standardize samples directory (#1727)** - **chore: standardize samples directory - delete unneeded dependencies (#1732)** - **feat: Add support for Python 3.12 (#1736)** - **feat: add `Client.query_and_wait` which directly returns a `RowIterator` of results (#1722)** - **chore: cleanup resources at startup (#1741)** - **docs: update multiple samples to change query to query_and_wait (#1784)** - **docs: update the query with no cache sample to use query_and_wait API (#1770)** - **docs: remove unused query code sample (#1769)** - **docs: Updates `query` to `query and wait` in samples/desktopapp/user_credentials.py (#1787)** - **docs: update `client_query_destination_table.py` sample to use `query_and_wait` (#1783)** - **docs: update to use `query_and_wait` in `client_query_w_positional_params.py` (#1786)** - **Docs: update the query destination table legacy file to use query_and_wait API (#1775)** - **docs: update sample for query_to_arrow to use query_and_wait API (#1776)** - **docs: update query_external_sheets_permanent_table.py to use query_and_wait API (#1778)** - **docs: update to use API (#1781)** - **docs: update to use `query_and_wait` in `samples/client_query_w_timestamp_params.py` (#1785)** - **docs: update to_geodataframe to use query_and_wait functionality (#1800)** - **docs: update `client_query_w_named_params.py` to use `query_and_wait` API (#1782)** - **docs(samples): updates to urllib3 constraint for Python 3.7 (#1834)** - **chore(deps): update all dependencies (#1640)** - **chore: use mock from unittest (#1823)** - **chore(deps): update all dependencies (#1835)** - **chore(deps): update all dependencies (#1849)** - **chore(deps): update dependency pyarrow to v15.0.2 (#1861)** - **fix: update error logging when converting to pyarrow column fails (#1836)** - **chore(deps): update dependency google-auth to v2.29.0 (#1865)** - **chore(deps): update dependency google-api-core to v2.18.0 (#1866)** - **feat: add `fields` parameter to `set_iam_policy` for consistency with update methods (#1872)** - **chore(deps): update all dependencies (#1873)** - **fix: make `pyarrow` an optional dependency post-3.20.0 yanked release (#1879)** - **chore(deps): update all dependencies (#1875)** - **chore(deps): update all dependencies (#1882)** - **chore(deps): update all dependencies (#1891)** - **chore(deps): update dependency idna to v3.7 [security] (#1896)** - **chore(deps): update all dependencies (#1893)** - **chore(deps): update all dependencies (#1904)** - **chore(deps): update all dependencies (#1916)** - **chore(deps): update all dependencies (#1921)** - **chore(deps): update all dependencies (#1926)** - **chore(deps): update all dependencies (#1934)** - **chore(deps): bump requests from 2.31.0 to 2.32.2 in /samples/geography (#1933)** - **chore(deps): update all dependencies (#1938)** - **testing: update BQML training option (#1943)** - **chore(deps): update all dependencies (#1946)** - **chore(deps): update all dependencies (#1954)** - **chore(deps): update dependency urllib3 to v2.2.2 [security] (#1959)** - **chore(deps): update all dependencies (#1955)** - **chore(deps): update dependency certifi to v2024.7.4 [security] (#1968)** - **feat: use `bigquery-magics` package for the `%%bigquery` magic (#1965)** - **docs: add short mode query sample & test (#1978)** - **chore(deps): update all dependencies (#1982)** - **chore(deps): update all dependencies (#1983)** - **chore(deps): update all dependencies (#1993)** - **chore(deps): update all dependencies (#2002)** - **chore(deps): update all dependencies (#2004)** - **testing: remove testing identity override (#2011)** - **chore(deps): update all dependencies (#2005)** - **chore(deps): update all dependencies (#2017)** - **chore(deps): update all dependencies (#2018)** - **chore(deps): update all dependencies (#2025)** - **chore(deps): update all dependencies (#2029)** - **chore(deps): bump fiona from 1.9.6 to 1.10.0 in /samples/geography (#2027)** - **chore(deps): update all dependencies (#2031)** - **chore(deps): update all dependencies (#2033)** - **build: use multiScm for Kokoro release builds (#2049)** - **chore(deps): update all dependencies (#2038)** - **chore(deps): update all dependencies (#2064)** - **chore(deps): update all dependencies (#2068)** - **chore(deps): update all dependencies (#2070)** - **chore(deps): update all dependencies (#2078)** - **chore(deps): update all dependencies (#2096)** - **chore(deps): update dependency grpcio to v1.69.0 (#2102)** - **test: add unit test covering the case where worker streams are stopped early (#2127)** - **deps: Remove Python 3.7 and 3.8 as supported runtimes (#2133)** - **chore(deps): update all dependencies (#2143)** - **chore(deps): update dependency pyasn1-modules to v0.4.2 (#2150)** - **chore(deps): update dependency shapely to v2.1.0 (#2155)** - **chore(deps): update dependency typing-extensions to v4.13.1 (#2156)** - **test: adds pytest-xdist to speed up processing of CI/CD checks (#2153)** - **chore(deps): update all dependencies (#2158)** - **chore(deps): update dependency db-dtypes to v1.4.3 (#2178)** - **chore(deps): update all dependencies to v3.32.0 (#2179)** - **fix(deps): update all dependencies (#2184)** - **chore(deps): update dependency google-auth to v2.40.2 (#2196)** - **docs: use query_and_wait in the array parameters sample (#2202)** - **feat: job creation mode GA (#2190)** - **chore(deps): update dependency pytest-xdist to v3.7.0 (#2203)** - **chore(deps): update all dependencies (#2205)** - **chore(deps): update all dependencies (#2209)** - **chore(deps): update all dependencies (#2211)** - **chore(deps): update dependency requests to v2.32.4 (#2213)** - **chore(deps): update dependency google-api-core to v2.25.1 (#2215)** - **chore(deps): bump urllib3 from 2.4.0 to 2.5.0 in /samples/geography (#2220)** - **chore: add label job sample (#2219)** - **chore(deps): update all dependencies (#2216)** - **chore(deps): update all dependencies (#2224)** - **chore(deps): update all dependencies (#2226)** - **chore(deps): update dependency certifi to v2025.7.9 (#2232)** - **chore(deps): update dependency certifi to v2025.7.14 (#2237)** - **chore(deps): update all dependencies to v3.35.0 (#2242)** - **chore(deps): update all dependencies (#2243)** - **chore(deps): update dependency matplotlib to v3.10.5 (#2251)** - **chore(deps): update all dependencies (#2253)** - **chore(deps): update all dependencies (#2259)** - **chore(deps): update all dependencies (#2275)** - **chore(deps): update all dependencies (#2300)** - **chore(deps): update dependency google-auth to v2.41.1 (#2312)** - **chore(deps): update all dependencies (#2314)** - **chore(python): Add Python 3.14 to python post processor image (#2317)** - **chore(deps): update dependency urllib3 to v2.6.0 [security] (#2342)** - **chore(deps): update dependency urllib3 to v2.6.3 [security] (#2357)** - **chore(deps): update dependency geopandas to v1.1.2 [security] (#2411)** - **chore(deps): update dependency pyasn1 to v0.6.2 [security] (#2407)** Fixes https://github.com/googleapis/google-cloud-python/issues/18150 --------- Signed-off-by: dependabot[bot] <support@github.com> Co-authored-by: WhiteSource Renovate <bot@renovateapp.com> Co-authored-by: Tim Swast <swast@google.com> Co-authored-by: gcf-owl-bot[bot] <78513119+gcf-owl-bot[bot]@users.noreply.github.com> Co-authored-by: Owl Bot <gcf-owl-bot[bot]@users.noreply.github.com> Co-authored-by: Bu Sun Kim <8822365+busunkim96@users.noreply.github.com> Co-authored-by: Steffany Brown <30247553+steffnay@users.noreply.github.com> Co-authored-by: john li <jli@circularly.org> Co-authored-by: meredithslota <meredithslota@google.com> Co-authored-by: Alma Becerril Salas <47731219+abecerrilsalas@users.noreply.github.com> Co-authored-by: Leah E. Cole <6719667+leahecole@users.noreply.github.com> Co-authored-by: Chalmer Lowe <chalmer.lowe@gmail.com> Co-authored-by: Dan Lee <71398022+dandhlee@users.noreply.github.com> Co-authored-by: aribray <45905583+aribray@users.noreply.github.com> Co-authored-by: Chalmer Lowe <chalmerlowe@google.com> Co-authored-by: Mario Torres Jr <105736410+Mattix23@users.noreply.github.com> Co-authored-by: Jared Chapman <thejaredchapman@gmail.com> Co-authored-by: adhiggs <38925938+adhiggs@users.noreply.github.com> Co-authored-by: Alvaro Viebrantz <aviebrantz@google.com> Co-authored-by: dependabot[bot] <49699333+dependabot[bot]@users.noreply.github.com> Co-authored-by: Lingqing Gan <lingqing.gan@gmail.com> Co-authored-by: Kira <kirnendra@google.com> Co-authored-by: Salem Jorden <115185670+SalemJorden@users.noreply.github.com> Co-authored-by: Salem Boyland <salemb@google.com> Co-authored-by: Stephanie A <129541811+DevStephanie@users.noreply.github.com> Co-authored-by: shollyman <shollyman@google.com> Co-authored-by: Leah Cole <coleleah@google.com>
FWIW the only real code changes occur in
gcloud/credentials.pygcloud/datastore/__init__.pygcloud/datastore/demo/__init__.pygcloud/storage/__init__.pygcloud/storage/connection.py-- Some helpers added here should probably be moved elsewhere. Maybe I file a bug?gcloud/storage/demo/__init__.pyregression/regression_utils.pyAlso note that googleapis/oauth2client#93 may be problematic for us, but only in the short term (on the order of days).
Also, I still need to update the encrypted environment variables on Travis.